)]}'
{"id":"openstack%2Fnetworking-bagpipe~841917","triplet_id":"openstack%2Fnetworking-bagpipe~master~Ia28bd3e4aef40cd490e4a98283b1b2507350f56f","project":"openstack/networking-bagpipe","branch":"master","topic":"drop-py3-6-7","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-07-26 11:10:28.000000000","reason":"Change was submitted"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-07-26 11:10:28.000000000","reason":"Change was submitted"},"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2022-07-26 11:10:28.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-05-26 15:44:47.000000000","reason":"Slawek Kaplonski replied on the change"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-07-26 09:32:22.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}},"hashtags":[],"change_id":"Ia28bd3e4aef40cd490e4a98283b1b2507350f56f","subject":"py310: Add required-projects list to py310 job","status":"MERGED","created":"2022-05-16 16:25:46.000000000","updated":"2022-07-26 11:12:16.000000000","submitted":"2022-07-26 11:10:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"841917-drop-py3-6-7","meta_rev_id":"a63bdb426a598261a25a9c89fb4d8cf7bf6e2bf5","_number":841917,"virtual_id_number":841917,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-26 11:10:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2022-07-21 10:44:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2022-07-26 09:32:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2022-07-26 09:32:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-16 17:40:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-17 09:55:21.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-05-17 09:55:21.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2022-05-17 09:55:21.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-05-17 09:55:21.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"576338ab976cc6fe875e3302b152f0b452c81f54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-05-16 16:25:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74e546e89b6e20da37a75ef711eb270d02f0f034","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 17:40:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d2f0f7ed7d84abc813235b24b57dede : SUCCESS in 11m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74d4cb5daca64a16a6b2f07c74a43391 : SUCCESS in 8m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/471cd424b3ae451bac3f1332731254a0 : SUCCESS in 10m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85b3b49aca3d4b9f94efca9108cf65a0 : SUCCESS in 11m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bb66009a2a54dd3b3cb6047bbd7ee86 : SUCCESS in 7m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d03718526c94b0a9ba9268d291f472a : FAILURE in 3m 30s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6747b6a0f8cc45a1bb5842975c16b2b9 : SUCCESS in 8m 01s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/7a9a6a47c693417fa68c7c84823e51c8 : SUCCESS in 55m 51s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/a3d780f69ff9409698ff193af8f153e3 : FAILURE in 9m 46s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/27b8878647344aef9a5996c8ecb265b5 : SUCCESS in 1h 10m 25s","accounts_in_message":[],"_revision_number":1},{"id":"33afab74c9582417e3c2d2d0013d6909a410578a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-05-17 09:59:23.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dadbaef68cc7833aa466a854ad30892d2bbf37bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-05-19 07:44:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"513c095654611269df91ee3d48782ff8d6576af0","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-05-19 07:56:40.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"712d81060c1abc4f785ceb3e9d3e85c02f4fdbb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 09:34:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76353b43cf36431b97e712199eccf28f : SUCCESS in 8m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d3ecd6d143f42d9bca60e6508a2ac03 : SUCCESS in 8m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01d6982e014345c3b449dae10dc51048 : SUCCESS in 10m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5fb9d25e1e4c45999b70f09dd97d2051 : SUCCESS in 10m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f1c5f9de32d4da6b2da52c913e546cb : SUCCESS in 8m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74567c46169a4441a3d07d368c93bf26 : FAILURE in 4m 08s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af69df36aec74801861d7666fc2faa1b : SUCCESS in 10m 00s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/4d445d141e574ca5abaf80f9e76dbc6c : SUCCESS in 1h 14m 40s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/73c94bddd6734e71ac856a2fc50cc4ab : FAILURE in 18m 47s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/b033ea09fcdc44f398edd308383c8749 : SUCCESS in 1h 44m 56s","accounts_in_message":[],"_revision_number":2},{"id":"e4679840ec01f32f019ab85736e71338dc4a6ee0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-05-20 07:15:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"db2649ea29f937165bf9c50a855a7612dcfee4e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 07:15:44.000000000","message":"Patch Set 3: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/networking-bagpipe on branch master.  The error was:\n\n  while scanning an alias\n  in \"openstack/networking-bagpipe/.zuul.yaml@master\", line 27, column 32\ndid not find expected alphabetic or numeric character\n  in \"openstack/networking-bagpipe/.zuul.yaml@master\", line 27, column 33","accounts_in_message":[],"_revision_number":3},{"id":"1bf6dada97d23d29e3f5b628b7b977f82776369a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-05-20 07:16:41.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"4a3a908707d8d237114af234ff1aff4750b2a6f5","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-05-20 07:17:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"102205acc13ade7fe9117d0697936336e6800860","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 07:17:13.000000000","message":"Patch Set 4: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/networking-bagpipe on branch master.  The error was:\n\n  while parsing a block mapping\n  in \"openstack/networking-bagpipe/.zuul.yaml@master\", line 27, column 13\ndid not find expected key\n  in \"openstack/networking-bagpipe/.zuul.yaml@master\", line 28, column 15","accounts_in_message":[],"_revision_number":4},{"id":"89d88e9ad7968b98dce5aadca9150b07b1c0647e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-05-20 09:52:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b5c93eb4111ddd1aae658fb235c443ef5d789faf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 09:52:58.000000000","message":"Patch Set 5: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/networking-bagpipe on branch master.  The error was:\n\n  found duplicate anchor; first occurrence\n  in \"openstack/networking-bagpipe/.zuul.yaml@master\", line 21, column 32\nsecond occurrence\n  in \"openstack/networking-bagpipe/.zuul.yaml@master\", line 47, column 32","accounts_in_message":[],"_revision_number":5},{"id":"d2a23af69d46aa04b25c92b086da0fd4fea5204a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-05-20 14:27:19.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"ef6df3456084674c52e335412e9d0683749d02cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 15:26:43.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3898bbc35a6449b8e4f046c533ffb4d : SUCCESS in 7m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad15b89860a4460c9223e422213e2e7d : SUCCESS in 8m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0328d00d335f43e992302bc5b8b1d55a : SUCCESS in 8m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e9e129d29ca14ccb9cf999111ca0eeb8 : SUCCESS in 8m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01532b725a194f058263ad5d5756ef27 : SUCCESS in 7m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2476d423e0f43cb9e71bc9ea2b50305 : FAILURE in 3m 30s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d23a83427cd4ddaa75f5ce8eb949fa2 : SUCCESS in 8m 42s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/5a63fa4e08564550891788a66a17fb5d : SUCCESS in 55m 12s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/d289958fb1974ecca6c9dfa502af174d : FAILURE in 9m 03s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/13e87619d59245089e1bb77fe432238a : SUCCESS in 58m 16s","accounts_in_message":[],"_revision_number":6},{"id":"9f22d1d0adffe53d318e42dcdba678b75fb92f5a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-05-26 15:44:47.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"70f76bd173c2ea47f7aed589e7642bcabf0a0e85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-07-18 11:40:10.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"8513332c3edd74ae44151caef99927f1ba94d380","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 13:28:37.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/69415257ade242aaab15b3f80ab36225\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2ac15913ec649fd95ce82df77501091 : SUCCESS in 8m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7024a0941def487097053392c183a251 : FAILURE in 8m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/52a4d19ee4824b7ba027c86ab7732ee6 : FAILURE in 8m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a2cd83e91f14a2ca867e84ecf51873e : FAILURE in 8m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20ed8c4fd274474c8a80b93ed068835c : SUCCESS in 8m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5417922bf8d0419fb61bef4c6fdc278a : FAILURE in 8m 08s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/7a4bdd2b71b14b9d84597b31e4472b28 : SUCCESS in 1h 22m 50s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/d61e88d500654df880d5f7ff149e6ea9 : FAILURE in 21m 17s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/52fbdc61a2be4324ac08e6ee90cde13c : SUCCESS in 1h 41m 20s","accounts_in_message":[],"_revision_number":7},{"id":"b078a10b4905271f1187a6b1421ad227166bd542","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-07-21 10:44:57.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"644c1a06c8f1525de6e0a52f75adea92d41ee3f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 12:16:43.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb612c94ed6a40c19f65f22660735f3b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc84f7d7d7f348b29cfc6819b55e02d2 : SUCCESS in 9m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d082caef03545b18543abdc62ec552e : SUCCESS in 11m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc37fa14e90c47cb95e2b61f151dd7cf : SUCCESS in 11m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d86c6e73db8e4abc9a9dac4afc4bd049 : SUCCESS in 10m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c77146792b34bc69f35bb2018d2819a : SUCCESS in 9m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4edc14fa2dc54ddcbefcb2fcff62b937 : SUCCESS in 10m 29s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/c26e35323bdf486f8decce2c0a77d059 : SUCCESS in 1h 23m 03s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/66b89f18f4074f61a77cb56b49f95779 : FAILURE in 19m 51s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/6fbdfbb927144e0982a262cd2e20ac3f : SUCCESS in 1h 25m 01s","accounts_in_message":[],"_revision_number":8},{"id":"6c74f3eed0dab59248bfb141c0ce2864df6de99c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-07-26 09:32:22.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"ec5bae0bce47356cba79f0508e4d29323df0c279","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 09:33:10.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"a32bea6a6d6cc7706bd4b8e40c2e53a075a809bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 11:10:27.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63970e56472e40d4859f391eed0e806b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/116166690f1347dd980c819a238d821c : SUCCESS in 10m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df3aaa9cb0c745518f562d7406cf87b4 : SUCCESS in 10m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14c31fba97da4917bc45e504844248c5 : SUCCESS in 10m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c6ac1cf51ca4dbca2e724a11657b80a : SUCCESS in 10m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e432da3f7f644505a837241c9816f2a3 : SUCCESS in 11m 20s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/b0763382ffc543b99d9830d6f9dd1d26 : SUCCESS in 1h 21m 29s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/543a07ee47da4632a99320ce92260b97 : SUCCESS in 1h 32m 55s","accounts_in_message":[],"_revision_number":8},{"id":"00cea4471fa32df0831d716c4fe9cd2eac85709a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 11:10:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"a63bdb426a598261a25a9c89fb4d8cf7bf6e2bf5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 11:12:16.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9724f568fc8042e28ed9420b23969c2c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1582fde827864801981f5e60f48660a9 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"9be4efd857a3501d37ebf3139f71244b2cdde927","revisions":{"90a8d834a59f96326d75392b86f8037342fcd4be":{"kind":"REWORK","_number":1,"created":"2022-05-16 16:25:46.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/17/841917/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/17/841917/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/1"}}},"commit":{"parents":[{"commit":"fd3f6d9c582052fbb4b25494d33525d7fe82ff0c","subject":"Update python testing as per zed cycle teting runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/fd3f6d9c582052fbb4b25494d33525d7fe82ff0c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-05-16 16:24:13.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-05-16 16:24:13.000000000","tz":120},"subject":"py310: Add required-projects list to py310 job","message":"py310: Add required-projects list to py310 job\n\nFor unit test jobs neutworking-bagpipe needs horizon, networking-bgpvpn\nand networking-sfc, so let\u0027s add these to py310 job also.\n\nChange-Id: Ia28bd3e4aef40cd490e4a98283b1b2507350f56f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/90a8d834a59f96326d75392b86f8037342fcd4be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/90a8d834a59f96326d75392b86f8037342fcd4be"}]},"branch":"refs/heads/master"},"441ea9d03dcf47e1b075b95dff28f1d9a0f5f12f":{"kind":"REWORK","_number":2,"created":"2022-05-19 07:44:13.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/17/841917/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/17/841917/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/2"}}},"commit":{"parents":[{"commit":"fd3f6d9c582052fbb4b25494d33525d7fe82ff0c","subject":"Update python testing as per zed cycle teting runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/fd3f6d9c582052fbb4b25494d33525d7fe82ff0c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-05-16 16:24:13.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-05-19 07:43:45.000000000","tz":120},"subject":"py310: Add required-projects list to py310 job","message":"py310: Add required-projects list to py310 job\n\nFor unit test jobs neutworking-bagpipe needs horizon, networking-bgpvpn\nand networking-sfc, so let\u0027s add these to py310 job also.\n\nChange-Id: Ia28bd3e4aef40cd490e4a98283b1b2507350f56f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/441ea9d03dcf47e1b075b95dff28f1d9a0f5f12f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/441ea9d03dcf47e1b075b95dff28f1d9a0f5f12f"}]},"branch":"refs/heads/master"},"b0966a96b7dfc22fe6f24b6104faec10b1bb5f35":{"kind":"REWORK","_number":3,"created":"2022-05-20 07:15:13.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/17/841917/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/17/841917/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/3"}}},"commit":{"parents":[{"commit":"fd3f6d9c582052fbb4b25494d33525d7fe82ff0c","subject":"Update python testing as per zed cycle teting runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/fd3f6d9c582052fbb4b25494d33525d7fe82ff0c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-05-16 16:24:13.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-05-20 07:14:54.000000000","tz":120},"subject":"py310: Add required-projects list to py310 job","message":"py310: Add required-projects list to py310 job\n\nFor unit test jobs neutworking-bagpipe needs horizon, networking-bgpvpn\nand networking-sfc, so let\u0027s add these to py310 job also.\n\nChange-Id: Ia28bd3e4aef40cd490e4a98283b1b2507350f56f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b0966a96b7dfc22fe6f24b6104faec10b1bb5f35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b0966a96b7dfc22fe6f24b6104faec10b1bb5f35"}]},"branch":"refs/heads/master"},"4e971d68cb85ce390ac5690869500e5fd5b0a5a6":{"kind":"REWORK","_number":4,"created":"2022-05-20 07:16:41.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/17/841917/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/17/841917/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/4"}}},"commit":{"parents":[{"commit":"fd3f6d9c582052fbb4b25494d33525d7fe82ff0c","subject":"Update python testing as per zed cycle teting runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/fd3f6d9c582052fbb4b25494d33525d7fe82ff0c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-05-16 16:24:13.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2022-05-20 07:16:41.000000000","tz":0},"subject":"py310: Add required-projects list to py310 job","message":"py310: Add required-projects list to py310 job\n\nFor unit test jobs neutworking-bagpipe needs horizon, networking-bgpvpn\nand networking-sfc, so let\u0027s add these to py310 job also.\n\nChange-Id: Ia28bd3e4aef40cd490e4a98283b1b2507350f56f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/4e971d68cb85ce390ac5690869500e5fd5b0a5a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/4e971d68cb85ce390ac5690869500e5fd5b0a5a6"}]},"branch":"refs/heads/master"},"7cbceb1d61837e7797fe01552f98bc6e4f98d8ab":{"kind":"REWORK","_number":5,"created":"2022-05-20 09:52:31.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/17/841917/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/17/841917/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/5"}}},"commit":{"parents":[{"commit":"529535ef83019ef67cc14f8af933fc93b43d89cd","subject":"Merge \"CI: add required projects for unit test jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/529535ef83019ef67cc14f8af933fc93b43d89cd"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-05-16 16:24:13.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-05-20 09:51:39.000000000","tz":120},"subject":"py310: Add required-projects list to py310 job","message":"py310: Add required-projects list to py310 job\n\nFor unit test jobs neutworking-bagpipe needs horizon, networking-bgpvpn\nand networking-sfc, so let\u0027s add these to py310 job also.\n\nChange-Id: Ia28bd3e4aef40cd490e4a98283b1b2507350f56f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7cbceb1d61837e7797fe01552f98bc6e4f98d8ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7cbceb1d61837e7797fe01552f98bc6e4f98d8ab"}]},"branch":"refs/heads/master"},"c7503c5f49bae868c4a0d7ed05ce947467e7d92b":{"kind":"REWORK","_number":6,"created":"2022-05-20 14:27:19.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/17/841917/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/17/841917/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/6"}}},"commit":{"parents":[{"commit":"529535ef83019ef67cc14f8af933fc93b43d89cd","subject":"Merge \"CI: add required projects for unit test jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/529535ef83019ef67cc14f8af933fc93b43d89cd"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-05-16 16:24:13.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2022-05-20 14:27:19.000000000","tz":0},"subject":"py310: Add required-projects list to py310 job","message":"py310: Add required-projects list to py310 job\n\nFor unit test jobs neutworking-bagpipe needs horizon, networking-bgpvpn\nand networking-sfc, so let\u0027s add these to py310 job also.\n\nChange-Id: Ia28bd3e4aef40cd490e4a98283b1b2507350f56f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c7503c5f49bae868c4a0d7ed05ce947467e7d92b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c7503c5f49bae868c4a0d7ed05ce947467e7d92b"}]},"branch":"refs/heads/master"},"7d5386cf0a154361b819a5af3f537920f7c28fcf":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2022-07-18 11:40:10.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/17/841917/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/17/841917/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/7"}}},"commit":{"parents":[{"commit":"d472fb7b5d05b70f9f4e12288eee1a9a01fdc9fd","subject":"Merge \"Drop lower-constraints.txt and its testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/d472fb7b5d05b70f9f4e12288eee1a9a01fdc9fd"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-05-16 16:24:13.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2022-07-18 11:40:10.000000000","tz":0},"subject":"py310: Add required-projects list to py310 job","message":"py310: Add required-projects list to py310 job\n\nFor unit test jobs neutworking-bagpipe needs horizon, networking-bgpvpn\nand networking-sfc, so let\u0027s add these to py310 job also.\n\nChange-Id: Ia28bd3e4aef40cd490e4a98283b1b2507350f56f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7d5386cf0a154361b819a5af3f537920f7c28fcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7d5386cf0a154361b819a5af3f537920f7c28fcf"}]},"branch":"refs/heads/master","description":"Rebase"},"9be4efd857a3501d37ebf3139f71244b2cdde927":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2022-07-21 10:44:57.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/17/841917/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/17/841917/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/17/841917/8"}}},"commit":{"parents":[{"commit":"428261187be105e8b966c22ee3af4f6a7234e663","subject":"Merge \"Py3: Import urllib properly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/428261187be105e8b966c22ee3af4f6a7234e663"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-05-16 16:24:13.000000000","tz":120},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2022-07-21 10:44:57.000000000","tz":0},"subject":"py310: Add required-projects list to py310 job","message":"py310: Add required-projects list to py310 job\n\nFor unit test jobs neutworking-bagpipe needs horizon, networking-bgpvpn\nand networking-sfc, so let\u0027s add these to py310 job also.\n\nChange-Id: Ia28bd3e4aef40cd490e4a98283b1b2507350f56f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/9be4efd857a3501d37ebf3139f71244b2cdde927"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/9be4efd857a3501d37ebf3139f71244b2cdde927"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
