)]}'
{"id":"openstack%2Fnetworking-bagpipe~842369","triplet_id":"openstack%2Fnetworking-bagpipe~master~I99f862e021fe8a93c3514010ebfa690f09ee54fd","project":"openstack/networking-bagpipe","branch":"master","topic":"bug/1973780","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-08-04 10:54:44.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-07-26 15:49:47.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-08-04 12:33:39.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-08-04 12:33:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I99f862e021fe8a93c3514010ebfa690f09ee54fd","subject":"Remove \"distutils\" library","status":"MERGED","created":"2022-05-18 14:39:54.000000000","updated":"2022-08-04 12:34:51.000000000","submitted":"2022-08-04 12:33:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842369-bug/1973780","meta_rev_id":"dc9cee85dedd16053100d990ee6ecd8388e21c4e","_number":842369,"virtual_id_number":842369,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-04 12:33:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-08-04 10:54:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-07-26 15:49:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-08-04 10:54:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-18 16:05:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-26 09:32:48.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-07-26 09:32:48.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-08-04 10:54:44.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"7d0d8eebedf287b91e8c4eab8f82cccbfbd2437e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-05-18 14:39:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1ef48cd489875c0a77f6436672545488390043a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 16:05:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/578e588b83d3424895d0b6efd7d57319 : SUCCESS in 10m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb5a2801f46245dfb7f413f0248f6a86 : SUCCESS in 10m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84ffb741477d49b6ae966f57fe916e6e : FAILURE in 7m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b78ae0fd1585429c83545c724ce334ce : FAILURE in 11m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cab25a7e4634a599b26768b5ae2f956 : SUCCESS in 9m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02fc8139d0224a0c8fff51beb6c1d8bb : FAILURE in 4m 38s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a3566412d534aa0b41402b6a1b4461e : FAILURE in 8m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93396fd850444da484dd9a9cb45a21b1 : SUCCESS in 9m 51s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/02b093fed4be462bbcb49d375e41bf3e : SUCCESS in 1h 14m 22s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/49c8b016277a4cde935491991a39548c : FAILURE in 17m 51s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/230c601fee1f4791a1a781304518963b : FAILURE in 1h 19m 44s","accounts_in_message":[],"_revision_number":1},{"id":"41aac2e7e838202fb32d6b125642f404946c1724","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-05-27 06:37:21.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c32e04e8e457ba3063606c9c5a042a1ff4c6fbd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 08:08:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c472a49ff5c4049b120ccd7172a568f : SUCCESS in 9m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83919cbe08634598baaca7f1ae5ad9bc : SUCCESS in 8m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c86f8232ab4f482a90f9fcadb5807ba4 : SUCCESS in 9m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e7c40caaea774c55956088fc338809d8 : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/550e56c7d0ec4ca382db3d0a8642c0ec : SUCCESS in 7m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39a81fcacca6467b83ac9e38a0abbfde : SUCCESS in 9m 25s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/abb7c9adbd6344e3a852f8644b4b07b5 : SUCCESS in 1h 25m 25s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/cfc703a9d906468881b3118eed036d1e : FAILURE in 20m 53s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/c9873d5097344c35b2e23065d2338ab1 : FAILURE in 1h 07m 28s","accounts_in_message":[],"_revision_number":2},{"id":"73cbbab3c6606850896075412b90864e9d8e3031","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-05-27 11:18:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4bdc516d665eade4f8e62f75f87afc9ba4eb4f1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 12:38:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8c1234c76d14bb3bce4651d473829c3 : SUCCESS in 9m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18fc642695fe448da517395f5e2a4999 : SUCCESS in 8m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/efa35f9b09be436eb0252fddb5806ae0 : SUCCESS in 8m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c60acbfce7e54282aa1b54c70559c215 : SUCCESS in 8m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df12ffbf727a4d9585d07da8c8fbce11 : SUCCESS in 8m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2ed610007a94989aa03fd9e57357ea7 : SUCCESS in 9m 46s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/0d05eaa4236c4b2c8e27a023edbec6f5 : SUCCESS in 1h 08m 03s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/fa0f2ed70a574a57aaa40cad3e86eac6 : FAILURE in 11m 38s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/c2479a94b67b4687b408e376d5b16bed : SUCCESS in 1h 17m 52s","accounts_in_message":[],"_revision_number":2},{"id":"3d4f41ad6e54ad3be8dcfed3ae0883100923d50f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-07-18 11:39:30.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"e16a7ec93a9eb5af346061c51f3d6c1c04bd4e3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-07-18 11:40:27.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"bbbc13b7c2ce4f5027e4511491e3fd3412e15d2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 13:32:09.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a6f035b19524f1e9a0c23a988818101\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b480ac5308df4e2c8a24a3b36aeacd47 : SUCCESS in 9m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63112cf63a00426c90ce06609712e1f6 : FAILURE in 8m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4097d7eab574bc98e5429c544f4aff7 : FAILURE in 10m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3dfb7846805a4f8f89509f0a312c8527 : FAILURE in 8m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5d4dad07e524b92aecf8cd718a2b139 : SUCCESS in 7m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/786f2386b93143d2b703c3644155173e : FAILURE in 8m 14s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/7b00ea3cb5024407b1b7f2295f6406fb : SUCCESS in 1h 21m 28s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/2a6ebd0182734d938debe11d8d7644a0 : FAILURE in 22m 57s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/c8b0d9788211487ba6f0535138e3b73f : SUCCESS in 1h 43m 57s","accounts_in_message":[],"_revision_number":4},{"id":"165e233433f2319fc12e78be701309c431ad509a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-07-21 10:45:03.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"772a10d97472a5108d846839d7242b7cd9f16b26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 12:15:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e43db1ee95bb4c67a29dbfac53a5b224\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a1489dae88f4fd7b57f185e93cccf17 : SUCCESS in 8m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f895bf3e67a41a4a1095981174d816a : SUCCESS in 11m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5430601564ac40fc9572d21db721c255 : SUCCESS in 12m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d2b8e0b735e42c8a7718ee5e2b2741e : SUCCESS in 12m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e929fb425beb4697ad3b22870806cd5d : SUCCESS in 12m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00fbbc9c2b12477db10bd26889eb0b0a : SUCCESS in 12m 46s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/16a58add67b94a328a44ce4188c37c37 : SUCCESS in 1h 18m 17s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/6b7248976d2648fe9d9fd1dfe0703d1a : FAILURE in 16m 11s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/1ed917bb0e084b179c466c0e871f5898 : SUCCESS in 1h 27m 24s","accounts_in_message":[],"_revision_number":5},{"id":"b21c7318a6ab657881e8621517746306466cbfa0","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-07-26 15:37:05.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"972e9f9264f62a7fbd733a26e5bf1ef32e76a630","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-07-26 15:49:47.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"1774eccd7385a2341c4b122e5be5a5c6418946d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 17:12:00.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d15e362e4be5409a8dcd10054596c6f7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d6cf6367f8545ef973f59debba65bbf : SUCCESS in 9m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ca7efa3a483466797887803dada5e1e : SUCCESS in 10m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5673cc6cabc4228a99c33d1160a491d : SUCCESS in 10m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/419d61989293409981dd281e213f3cfd : SUCCESS in 9m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bf12629ec2547b0b40672b747e38565 : SUCCESS in 8m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76e5544a0c1b463a955f76533a645af8 : SUCCESS in 10m 23s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/5735f02f767146538f039997d5ecef7a : SUCCESS in 1h 20m 33s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/ceaa85b01560420eba7b81830d4a5af5 : FAILURE in 16m 29s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/54086ec1ee464cf489e1b6b7762d57f7 : SUCCESS in 1h 32m 26s","accounts_in_message":[],"_revision_number":5},{"id":"c8c0036c8e24f9e833f61a9091b552ca6aa87256","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-04 10:54:44.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"423e1a3150c02cf0492a82468b4c881d742fe6e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 10:56:51.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"487060c4b6e271611af57284acbd9e23163e8cab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 12:33:38.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1f4000ecd7842b190cce03c1df2639c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f6cbb4482374485b411d4cf76b65f27 : SUCCESS in 8m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/231b0c555c4642bf879e2c83ba947057 : SUCCESS in 9m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/969034c65c6b4ee5a032a3403319487e : SUCCESS in 9m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/801333318b2c4d2284f5884b0e0efbb2 : SUCCESS in 7m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d77ef6ce256a426e8ff98826bfb23b46 : SUCCESS in 8m 13s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/dd090d2097884de88024c1a8d5a06a32 : SUCCESS in 1h 27m 29s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/c0ca728f97b64f96a0ba210e23d9fc43 : SUCCESS in 1h 30m 38s","accounts_in_message":[],"_revision_number":5},{"id":"586cef8e65b233264de2a0fe8a1e03a93c12cb15","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 12:33:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"dc9cee85dedd16053100d990ee6ecd8388e21c4e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 12:34:51.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c0d1e4546a34e7cbb95483941ef6a86\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5faa0a7ec50840cf8067465d26f41213 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"8e6051947150755c1f54fd06bafb4b5960481ea4","revisions":{"dd842f476ec037d8811e0574125eb9583c99a430":{"kind":"REWORK","_number":1,"created":"2022-05-18 14:39:54.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/69/842369/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/69/842369/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/69/842369/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/69/842369/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/69/842369/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/69/842369/1"}}},"commit":{"parents":[{"commit":"85d6c25c2f2f3a0434396c30b120189214dfc308","subject":"Add weekly jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/85d6c25c2f2f3a0434396c30b120189214dfc308"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-05-18 14:36:59.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-05-18 14:36:59.000000000","tz":120},"subject":"Remove \"distutils\" library","message":"Remove \"distutils\" library\n\nLibrary \"distutils\" will be marked as deprecated in Python 3.10:\nhttps://peps.python.org/pep-0386/\n\nThis patch does the following replacements, that provide the same\nfunctionality and API:\n- distutils.version.StrictVersion -\u003e versionutils.\nconvert_version_to_tuple\n- distutils.spawn.find_executable -\u003e shutil.which\n\nChange-Id: I99f862e021fe8a93c3514010ebfa690f09ee54fd\nCloses-Bug: #1973780\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/dd842f476ec037d8811e0574125eb9583c99a430"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/dd842f476ec037d8811e0574125eb9583c99a430"}]},"branch":"refs/heads/master"},"ab408eccfc89f0bd079ad320f87d8e4f0d460377":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-05-27 06:37:21.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/69/842369/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/69/842369/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/69/842369/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/69/842369/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/69/842369/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/69/842369/2"}}},"commit":{"parents":[{"commit":"c7503c5f49bae868c4a0d7ed05ce947467e7d92b","subject":"py310: Add required-projects list to py310 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c7503c5f49bae868c4a0d7ed05ce947467e7d92b"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-05-18 14:36:59.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2022-05-27 06:37:21.000000000","tz":0},"subject":"Remove \"distutils\" library","message":"Remove \"distutils\" library\n\nLibrary \"distutils\" will be marked as deprecated in Python 3.10:\nhttps://peps.python.org/pep-0386/\n\nThis patch does the following replacements, that provide the same\nfunctionality and API:\n- distutils.version.StrictVersion -\u003e versionutils.\nconvert_version_to_tuple\n- distutils.spawn.find_executable -\u003e shutil.which\n\nChange-Id: I99f862e021fe8a93c3514010ebfa690f09ee54fd\nCloses-Bug: #1973780\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/ab408eccfc89f0bd079ad320f87d8e4f0d460377"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/ab408eccfc89f0bd079ad320f87d8e4f0d460377"}]},"branch":"refs/heads/master","description":"Rebase"},"71d6764462e080a4d7252845b557fef41e9fc411":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-07-18 11:39:30.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/69/842369/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/69/842369/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/69/842369/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/69/842369/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/69/842369/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/69/842369/3"}}},"commit":{"parents":[{"commit":"d472fb7b5d05b70f9f4e12288eee1a9a01fdc9fd","subject":"Merge \"Drop lower-constraints.txt and its testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/d472fb7b5d05b70f9f4e12288eee1a9a01fdc9fd"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-05-18 14:36:59.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2022-07-18 11:39:30.000000000","tz":0},"subject":"Remove \"distutils\" library","message":"Remove \"distutils\" library\n\nLibrary \"distutils\" will be marked as deprecated in Python 3.10:\nhttps://peps.python.org/pep-0386/\n\nThis patch does the following replacements, that provide the same\nfunctionality and API:\n- distutils.version.StrictVersion -\u003e versionutils.\nconvert_version_to_tuple\n- distutils.spawn.find_executable -\u003e shutil.which\n\nChange-Id: I99f862e021fe8a93c3514010ebfa690f09ee54fd\nCloses-Bug: #1973780\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/71d6764462e080a4d7252845b557fef41e9fc411"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/71d6764462e080a4d7252845b557fef41e9fc411"}]},"branch":"refs/heads/master","description":"Rebase"},"2633d945ecb4d43eb85048b5b73d11e036ce924c":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-07-18 11:40:27.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/69/842369/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/69/842369/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/69/842369/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/69/842369/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/69/842369/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/69/842369/4"}}},"commit":{"parents":[{"commit":"7d5386cf0a154361b819a5af3f537920f7c28fcf","subject":"py310: Add required-projects list to py310 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7d5386cf0a154361b819a5af3f537920f7c28fcf"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-05-18 14:36:59.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2022-07-18 11:40:27.000000000","tz":0},"subject":"Remove \"distutils\" library","message":"Remove \"distutils\" library\n\nLibrary \"distutils\" will be marked as deprecated in Python 3.10:\nhttps://peps.python.org/pep-0386/\n\nThis patch does the following replacements, that provide the same\nfunctionality and API:\n- distutils.version.StrictVersion -\u003e versionutils.\nconvert_version_to_tuple\n- distutils.spawn.find_executable -\u003e shutil.which\n\nChange-Id: I99f862e021fe8a93c3514010ebfa690f09ee54fd\nCloses-Bug: #1973780\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/2633d945ecb4d43eb85048b5b73d11e036ce924c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/2633d945ecb4d43eb85048b5b73d11e036ce924c"}]},"branch":"refs/heads/master","description":"Rebase"},"8e6051947150755c1f54fd06bafb4b5960481ea4":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-07-21 10:45:03.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/69/842369/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/69/842369/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/69/842369/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/69/842369/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/69/842369/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/69/842369/5"}}},"commit":{"parents":[{"commit":"9be4efd857a3501d37ebf3139f71244b2cdde927","subject":"py310: Add required-projects list to py310 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/9be4efd857a3501d37ebf3139f71244b2cdde927"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-05-18 14:36:59.000000000","tz":120},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2022-07-21 10:45:03.000000000","tz":0},"subject":"Remove \"distutils\" library","message":"Remove \"distutils\" library\n\nLibrary \"distutils\" will be marked as deprecated in Python 3.10:\nhttps://peps.python.org/pep-0386/\n\nThis patch does the following replacements, that provide the same\nfunctionality and API:\n- distutils.version.StrictVersion -\u003e versionutils.\nconvert_version_to_tuple\n- distutils.spawn.find_executable -\u003e shutil.which\n\nChange-Id: I99f862e021fe8a93c3514010ebfa690f09ee54fd\nCloses-Bug: #1973780\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/8e6051947150755c1f54fd06bafb4b5960481ea4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/8e6051947150755c1f54fd06bafb4b5960481ea4"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
