)]}'
{"id":"openstack%2Fnetworking-bagpipe~850348","triplet_id":"openstack%2Fnetworking-bagpipe~master~I045c03b0ae54a3620a91a0d88fdf2098607d0fa4","project":"openstack/networking-bagpipe","branch":"master","topic":"not-supported-framework","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-07-19 15:45:43.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I045c03b0ae54a3620a91a0d88fdf2098607d0fa4","subject":"Use Linux Bridge experimental flag","status":"ABANDONED","created":"2022-07-19 12:55:26.000000000","updated":"2022-07-19 15:45:43.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a288c014b094551f242fad408fc9b46e898facd6","_number":850348,"virtual_id_number":850348,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-07-19 14:20:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-19 14:20:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1dcf7341915faa59307c56e94d3db60458d5d277","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-07-19 12:55:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23d090ebaa3db2d616728f4517caa756ad8f677f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 14:20:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c5121f97aa2454ca33f98c6a6900d6c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac31905b29d0449c8acdfff92ab707a4 : SUCCESS in 9m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3b3115632c94bdc8473a65906e74a8d : FAILURE in 8m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0581b130b2bd444d9cf228bdb000d471 : FAILURE in 10m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8ce8e3dd8774f0e9f668d83f55cd896 : FAILURE in 9m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36ca2877b75540d58fbaf68cd365d0ea : SUCCESS in 9m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24311bfa0f2a40bca244dae062bc2555 : FAILURE in 9m 13s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/2af2e075a58f47a4a9fb933d65d82a2a : SUCCESS in 1h 15m 00s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/1cab54fafbe041a3abb383d1c2eee630 : FAILURE in 15m 47s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/61a294b9e7a4425e9d4243c86c997441 : SUCCESS in 1h 23m 27s","accounts_in_message":[],"_revision_number":1},{"id":"a288c014b094551f242fad408fc9b46e898facd6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-07-19 15:45:43.000000000","message":"Abandoned\n\nSuperseded by https://review.opendev.org/c/openstack/networking-bagpipe/+/850340","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"15270efb319859dd17e38cce9db12d8bb2b596d4","revisions":{"15270efb319859dd17e38cce9db12d8bb2b596d4":{"kind":"REWORK","_number":1,"created":"2022-07-19 12:55:26.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/48/850348/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/48/850348/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/850348/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/850348/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/850348/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/48/850348/1"}}},"commit":{"parents":[{"commit":"d472fb7b5d05b70f9f4e12288eee1a9a01fdc9fd","subject":"Merge \"Drop lower-constraints.txt and its testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/d472fb7b5d05b70f9f4e12288eee1a9a01fdc9fd"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-07-09 15:35:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-07-09 15:37:32.000000000","tz":0},"subject":"Use Linux Bridge experimental flag","message":"Use Linux Bridge experimental flag\n\nSince [1], Linux Bridge mech driver has been marked as experimental\n(unsupported). Please check [1] for more information.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/845181\n[2]https://lists.openstack.org/pipermail/openstack-discuss/2022-April/028164.html\n\nChange-Id: I045c03b0ae54a3620a91a0d88fdf2098607d0fa4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/15270efb319859dd17e38cce9db12d8bb2b596d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/15270efb319859dd17e38cce9db12d8bb2b596d4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
