)]}'
{"id":"openstack%2Fnetworking-bagpipe~851159","triplet_id":"openstack%2Fnetworking-bagpipe~master~I63c30f2421c82d8e8c8bb2af4a373e11a9a4968f","project":"openstack/networking-bagpipe","branch":"master","topic":"dnm","hashtags":[],"change_id":"I63c30f2421c82d8e8c8bb2af4a373e11a9a4968f","subject":"DNM: test master","status":"ABANDONED","created":"2022-07-27 07:10:28.000000000","updated":"2022-07-27 09:20:30.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"44566660e02ca7053a8988149bf530ebfa3e245f","_number":851159,"virtual_id_number":851159,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-07-27 08:38:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-27 08:38:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"993bce06f1783367cb37301519af07f195ab511a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-07-27 07:10:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08be8710aa13a3ca7b13657a3180a5a425b72257","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 08:38:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c46a0e2728454f73a8a6b1e23f6b9e35\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b81eaf54543e43ed8f40407369782925 : SUCCESS in 8m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d5265efc6a34f8ba6a7fd9623d3b159 : SUCCESS in 7m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9900b6a86df945aabe60319ee0b86e1e : SUCCESS in 8m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b5ec2796b7a848ce93fa247544ea93c7 : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66fd5365bd654deea641ae2e7ce90fe3 : SUCCESS in 7m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73c4d29df8f942f0a2414936b7d6e689 : SUCCESS in 8m 58s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/d2a1661fe4c64c59b1506d9cb5502246 : SUCCESS in 1h 14m 33s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/e1d074870c594f83b12deaf8a8997ee0 : FAILURE in 13m 10s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/0aa13883d23f4d5385c8f83d7347b52a : SUCCESS in 1h 27m 24s","accounts_in_message":[],"_revision_number":1},{"id":"44566660e02ca7053a8988149bf530ebfa3e245f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-07-27 09:20:30.000000000","message":"Abandoned\n\nall seems ok, the periodic failure was a gap most probably","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f4241671cb56cf971e6c9df1bcd0b394977c8cd2","revisions":{"f4241671cb56cf971e6c9df1bcd0b394977c8cd2":{"kind":"REWORK","_number":1,"created":"2022-07-27 07:10:28.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/59/851159/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/59/851159/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/59/851159/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/59/851159/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/59/851159/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/59/851159/1"}}},"commit":{"parents":[{"commit":"9be4efd857a3501d37ebf3139f71244b2cdde927","subject":"py310: Add required-projects list to py310 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/9be4efd857a3501d37ebf3139f71244b2cdde927"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-07-27 07:09:56.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-07-27 07:09:56.000000000","tz":120},"subject":"DNM: test master","message":"DNM: test master\n\nChange-Id: I63c30f2421c82d8e8c8bb2af4a373e11a9a4968f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f4241671cb56cf971e6c9df1bcd0b394977c8cd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f4241671cb56cf971e6c9df1bcd0b394977c8cd2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
