)]}'
{"id":"openstack%2Fnetworking-bagpipe~861214","triplet_id":"openstack%2Fnetworking-bagpipe~stable%2Fzed~Iceb92b157d5e875090a0dedec37c398716b89495","project":"openstack/networking-bagpipe","branch":"stable/zed","topic":"add-bagpipe-required-projects","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-10-21 09:56:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iceb92b157d5e875090a0dedec37c398716b89495","subject":"Add required projects where necessary","status":"MERGED","created":"2022-10-13 18:35:21.000000000","updated":"2022-10-21 09:58:11.000000000","submitted":"2022-10-21 09:56:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"861214-add-bagpipe-required-projects","meta_rev_id":"38ded32682d349cb17e3c2a7b08ed3074eb4f686","_number":861214,"virtual_id_number":861214,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-21 09:56:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-10-21 08:22:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-10-14 07:28:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-10-21 08:22:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-13 20:11:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-14 07:28:29.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-10-21 08:22:48.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"7605a78387dcc0b419cae5b32528a80676685982","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-10-13 18:35:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4aaff32f371a2dd4ff257fd4cd91e8908f95e256","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-10-13 18:54:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"039dde67d8627ce87ef20f5a3e8ac6c7023b207b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-10-13 19:07:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"313946338d68675e2765347324f1217e82f32734","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 20:11:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ce286628aca45718e1424d241fb341d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaf17f0946a8494483d39f6619fb5c9b : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6af1e6ec7e124aee9cf9d2d35718128f : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59e26501e0d14074b17ac89c14a319e1 : SUCCESS in 6m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cbd251dd9c81454d8fe07d44ed43a59e : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/318c37f557fe407a87550672e13cab96 : SUCCESS in 7m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c8684f407d248aba702aff0f48c09ac : SUCCESS in 7m 11s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/944dcc5223814c69833285ea2a5eeff7 : SUCCESS in 58m 01s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/27c69bf7a3604d6caeeb7ea6dd6da562 : FAILURE in 9m 06s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/62e3deb25d964af49b703fda021eb340 : FAILURE in 1h 12m 03s","accounts_in_message":[],"_revision_number":2},{"id":"2d9dd2c14caac7b5798e734453c21ba9c1eb038e","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-10-14 05:07:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b773461712365b5f23a02c7e6171240a3536306c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 06:40:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6af6da5e961b4f0a866c638aea85c8a6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/371dad0d1ed74d8aad86d9bd11166e82 : SUCCESS in 7m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50cc943e077a4c84b7be304e18332e82 : SUCCESS in 7m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93f4237224d942aa97f16f587098a1a1 : SUCCESS in 7m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/82a1da9c3f04443abc2cdaa9d649980c : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abceaaec12904444a4c43da92f834f84 : SUCCESS in 10m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd588fb657b4407ca4338a5a93c3f238 : SUCCESS in 7m 56s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/329450cd4f1547e3ab364dda0d8b81b9 : SUCCESS in 1h 18m 01s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/ef1dbfcfe3664056a7e7b0ebb4eda76f : FAILURE in 19m 00s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/f8667d4d66ea42e8a0da19798fa33b49 : SUCCESS in 1h 27m 56s","accounts_in_message":[],"_revision_number":2},{"id":"645c90c4265b55a1b010d4934cf2b2dc431912fa","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-10-14 07:28:29.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"04e7b32550772cbcb9f3b7648d3c0bcb7fe18d3f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-10-21 08:22:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9ad17708e4a732503ec460833c229008a9a10b02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 08:23:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"40367fd9859465916dbb737a0f853b80af94edfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 09:56:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdc64b20033346c197cc5df9fa44a01d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79f6590410d6450abe82ddba4d420717 : SUCCESS in 8m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fcb76114ec44d8a8c8d3544e68866d2 : SUCCESS in 7m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f90a3f475d894f20af8be5d1dfd72f2e : SUCCESS in 8m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9efe76232d3e4d4bbad79b1a66415552 : SUCCESS in 9m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/303ed5c213cb4bbfb5a4fe118d2de209 : SUCCESS in 8m 20s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/bce94679a13a4f7c8830aa1241d1ba31 : SUCCESS in 1h 20m 46s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/13af5c3a3ad4445a81fa034881c35230 : SUCCESS in 1h 26m 49s","accounts_in_message":[],"_revision_number":2},{"id":"7dce6fcff067ac41402d814eacf1e25b24133f83","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 09:56:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"38ded32682d349cb17e3c2a7b08ed3074eb4f686","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 09:58:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/321b299454e84a348229e7ea27e00660\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/142616fe1484493899d56235009965f9 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1b474c3333b25970431ade31d90a38af3f47fb7a","revisions":{"7df44e27df9682d7d5652962f0d5c4676a8d4cb6":{"kind":"REWORK","_number":1,"created":"2022-10-13 18:35:21.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/14/861214/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/14/861214/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/14/861214/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/14/861214/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/14/861214/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/14/861214/1"}}},"commit":{"parents":[{"commit":"1cfb16da8423bff87535f62060985942e25eef73","subject":"Update TOX_CONSTRAINTS_FILE for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/1cfb16da8423bff87535f62060985942e25eef73"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-10-13 18:23:10.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-10-13 18:33:50.000000000","tz":120},"subject":"Add required projects where necessary","message":"Add required projects where necessary\n\nMost of the jobs needs required-projects to be added, especially in\nstable branches, where the lack of them causes gate failures.\n\nChange-Id: Iceb92b157d5e875090a0dedec37c398716b89495\n(cherry picked from commit f5f58021001bcdd078ba65d8d0fa75720caef529)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7df44e27df9682d7d5652962f0d5c4676a8d4cb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7df44e27df9682d7d5652962f0d5c4676a8d4cb6"}]},"branch":"refs/heads/stable/zed"},"1b474c3333b25970431ade31d90a38af3f47fb7a":{"kind":"REWORK","_number":2,"created":"2022-10-13 18:54:11.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/14/861214/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/14/861214/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/14/861214/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/14/861214/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/14/861214/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/14/861214/2"}}},"commit":{"parents":[{"commit":"1cfb16da8423bff87535f62060985942e25eef73","subject":"Update TOX_CONSTRAINTS_FILE for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/1cfb16da8423bff87535f62060985942e25eef73"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-10-13 18:23:10.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-10-13 18:52:28.000000000","tz":120},"subject":"Add required projects where necessary","message":"Add required projects where necessary\n\nMost of the jobs needs required-projects to be added, especially in\nstable branches, where the lack of them causes gate failures.\n\nSince the templates openstack-python3-jobs-neutron and\nperiodic-stable-jobs-neutron don\u0027t contain all the necessary projects\nfor bagpipe we need to add them one by one.\n\nNOTE(elod.illes): py310 is not part of stable/zed, so that is removed\nfrom this cherry pick. In contrary, py38 is part of stable/zed periodic\njob template, so it is added.\n\nChange-Id: Iceb92b157d5e875090a0dedec37c398716b89495\n(cherry picked from commit 7a02ee3188a4cadba76d75ea03400b35a33c2e5f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/1b474c3333b25970431ade31d90a38af3f47fb7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/1b474c3333b25970431ade31d90a38af3f47fb7a"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
