)]}'
{"id":"openstack%2Fnetworking-bagpipe~862288","triplet_id":"openstack%2Fnetworking-bagpipe~stable%2Fxena~Iceb92b157d5e875090a0dedec37c398716b89495","project":"openstack/networking-bagpipe","branch":"stable/xena","topic":"add-bagpipe-required-projects","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-11-09 18:42:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iceb92b157d5e875090a0dedec37c398716b89495","subject":"Add required projects where necessary","status":"MERGED","created":"2022-10-21 10:02:11.000000000","updated":"2022-11-09 18:44:50.000000000","submitted":"2022-11-09 18:42:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862288-add-bagpipe-required-projects","meta_rev_id":"4382c1d0803d193c8b75f51f360dac79b05dd6b7","_number":862288,"virtual_id_number":862288,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-09 18:42:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-11-09 17:21:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-10-25 05:19:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2022-11-09 17:21:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-21 11:36:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-25 05:19:41.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-11-09 17:21:38.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"cd8a44a4106bf51e041c9ce519569350e6d454a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-10-21 10:02:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"72a671419af96e7371e13f6032681da1f3db3491","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 11:36:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b65f9531afd49538d301d629a569e51\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4718fb7994a344d5b7035074efcda682 : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b252b95a42d9435dbbf06e805c7a4ca6 : SUCCESS in 8m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bd73e0357854e60b3af271214b87a42 : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/123632135e5f4b29839eaa2883f88f37 : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/affa89d45ce64205806d1c6aa0f64797 : SUCCESS in 8m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/113162f4ae724163a68c9af88c16264b : FAILURE in 2m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9674acda04bd4b19b7212798ab4d9a8e : SUCCESS in 8m 29s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/98d9ed2d320c45d9bb11fe5f52485473 : SUCCESS in 1h 14m 30s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/bb70d7bbe84347f89533d78afd2020b8 : FAILURE in 7m 47s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/23a4dbf152a446ada2cb6e74e91e6276 : SUCCESS in 1h 27m 46s","accounts_in_message":[],"_revision_number":1},{"id":"5c38c2054fbd7cfd3228e83797a3a42887acd385","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-10-25 05:19:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"672678c7b1a3db8cf0e1ce39abfa5b9c3cbe0cad","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-11-09 17:21:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"82850da06068f64e977e2714f0fec2445c002df0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 17:22:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d152435dd44f364c03150d0241f256a788a67614","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 18:42:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5945db40cad344f8a8a3dd30cfb4804c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/667e1c608e834534a48536672550c555 : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1b47bfd361143d8bbd8e26d382cc212 : SUCCESS in 9m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b76b204f9494304a63fd5d648ddd1e1 : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26e26cefbc8e437ab86b6cc30cf01ad1 : SUCCESS in 10m 14s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/9c96d39204e24b72a822e30922186f89 : SUCCESS in 1h 07m 16s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/0c1ddee3abae493291741560ac921ea7 : SUCCESS in 1h 18m 20s","accounts_in_message":[],"_revision_number":1},{"id":"3439e1465eb5ab20413cb51e4d4ccacba4a3a9b4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 18:42:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4382c1d0803d193c8b75f51f360dac79b05dd6b7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 18:44:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f88509988a364ec293d8db991c6715b9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a36745bf597d40128a61c09f06f5a16b : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"39546f061193de8d46d6ca6141122f83bb18621a","revisions":{"39546f061193de8d46d6ca6141122f83bb18621a":{"kind":"REWORK","_number":1,"created":"2022-10-21 10:02:11.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/88/862288/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/88/862288/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/88/862288/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/88/862288/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/88/862288/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/88/862288/1"}}},"commit":{"parents":[{"commit":"fb63c4514a6d280e3692ae987c9ae15d7a3821bd","subject":"Update TOX_CONSTRAINTS_FILE for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/fb63c4514a6d280e3692ae987c9ae15d7a3821bd"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-10-13 18:23:10.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-10-21 09:59:36.000000000","tz":120},"subject":"Add required projects where necessary","message":"Add required projects where necessary\n\nMost of the jobs needs required-projects to be added, especially in\nstable branches, where the lack of them causes gate failures.\n\nSince the templates openstack-python3-jobs-neutron and\nperiodic-stable-jobs-neutron don\u0027t contain all the necessary projects\nfor bagpipe we need to add them one by one.\n\nChanges:\n  .zuul.yaml\n\nNOTE(elod.illes): change is due to py39 job is only present in the\ncheck queue of stable/xena. No need to add in the gate queue.\n\nChange-Id: Iceb92b157d5e875090a0dedec37c398716b89495\n(cherry picked from commit 7a02ee3188a4cadba76d75ea03400b35a33c2e5f)\n(cherry picked from commit 1b474c3333b25970431ade31d90a38af3f47fb7a)\n(cherry picked from commit d9bd4c660ee10c8de39d4b83090aefb79933fc00)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/39546f061193de8d46d6ca6141122f83bb18621a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/39546f061193de8d46d6ca6141122f83bb18621a"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
