)]}'
{"id":"openstack%2Fnetworking-bagpipe~862502","triplet_id":"openstack%2Fnetworking-bagpipe~stable%2Fwallaby~Iceb92b157d5e875090a0dedec37c398716b89495","project":"openstack/networking-bagpipe","branch":"stable/wallaby","topic":"add-bagpipe-required-projects","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-11-10 15:12:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iceb92b157d5e875090a0dedec37c398716b89495","subject":"Add required project and follow pyroute2 changes","status":"MERGED","created":"2022-10-24 16:13:48.000000000","updated":"2022-11-10 15:14:45.000000000","submitted":"2022-11-10 15:12:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862502-add-bagpipe-required-projects","meta_rev_id":"e927a8ce688abe4c7088ad93ab45503ef6240c29","_number":862502,"virtual_id_number":862502,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-10 15:12:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-11-10 13:34:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-11-10 13:04:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-11-10 13:34:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-11-10 10:50:15.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-24 17:23:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-09 17:24:55.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-11-10 10:38:34.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2022-11-10 10:50:15.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-11-10 13:04:21.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"55e7ce904c494e687fefa844692d6a696c65af98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-10-24 16:13:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea0f7a6d282101ad38671e08f6cde8c5a1b4e3e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 17:23:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/466fd49f4af943f8881e4d7d7b30c4a5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cad25e1582f48f889fea650e9fd2827 : FAILURE in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb9033c18ee14029a81c4e482d7e1c69 : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12f20f07785d42b19c6d455fe22231c4 : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc165b8b31a3469f9b586edcedcc04cb : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb3392e732d14811910dbee095e2e1c0 : SUCCESS in 8m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b180c50830e455a83a1cbcc26ae19a8 : FAILURE in 3m 02s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c05dbb8b8aa4308b680abd54f1295b1 : SUCCESS in 8m 50s\n- neutron-tempest-plugin-bgpvpn-bagpipe-wallaby https://zuul.opendev.org/t/openstack/build/1ef269ab053743a29c5359e17c40c6ff : SUCCESS in 57m 34s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/790154e2d3344f2aac289ab45b2ebfd2 : FAILURE in 7m 44s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/a7c0dba1ce1543449751232c57402445 : SUCCESS in 1h 07m 40s\n\nWarning:\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/vpn/evpn/linux_vxlan.py\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/vpn/ipvpn/mpls_linux_dataplane.py","accounts_in_message":[],"_revision_number":1},{"id":"98f6b2bf6afcd2618ebb5674d554a3c61e9eea20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-10-28 15:45:18.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"f16fbdfe138d2d96edfde6a6bcdb74f61300c3d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 17:09:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/409de424d82043118a585b4d8bea5ce4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aaa822818d24564816d5aa5d466fe1d : SUCCESS in 8m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af62eb427627433d903c989602d1dd52 : SUCCESS in 9m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b11cbe6896f401cb7a9ee2b0708949d : SUCCESS in 8m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8dc5dde1b9e4bbcbf60c1f7494b6e0b : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d74b8aa098b14463b4560b0801228678 : SUCCESS in 11m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c484053d12c6426e9b190e7a6a99de2b : FAILURE in 3m 30s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1521ff40bb6e4d44804a1c9cf25108e4 : SUCCESS in 7m 51s\n- neutron-tempest-plugin-bgpvpn-bagpipe-wallaby https://zuul.opendev.org/t/openstack/build/04dc62c593b745aab8f051d258fbe172 : SUCCESS in 1h 08m 29s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/ae1f536a43b14f389343611db954b791 : FAILURE in 10m 05s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/3d036166fd404d28b828936d5814e81c : SUCCESS in 1h 22m 26s","accounts_in_message":[],"_revision_number":2},{"id":"3862ce2107d696b7fdde1e53d74618667cfafd45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-11-09 17:24:05.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"a4e6c4df1b7670c74517e010c308fe76ff2acb69","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-11-09 17:24:55.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0f12932388158b1ea794f4e38e9dccf0eab89fa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 19:32:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97fec309592a4526a6fa2b512fd25534\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e720901f80e542ba8eed1974722eb953 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a1953d4e1bc4b1db03f418eb045b08c : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80459bbeb492443d8b48ecb76ad38925 : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6f49b1efd1541f792c40140f35c71d1 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87a1a8d09e0342ecbe94917627e011e8 : SUCCESS in 8m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9fd915d1775b4b19accd4fb393e3437f : FAILURE in 3m 28s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45a286bd83414b91867ebead6ba4923a : SUCCESS in 10m 52s\n- neutron-tempest-plugin-bgpvpn-bagpipe-wallaby https://zuul.opendev.org/t/openstack/build/8f7037a41a3043058e392ff9c742aed3 : SUCCESS in 2h 01m 37s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/283eef7a0a394d2f8bf2b0e7ec898f01 : FAILURE in 18m 45s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/6a651fb5cfaf4938bc04b1d6e7cf4638 : SUCCESS in 1h 42m 12s","accounts_in_message":[],"_revision_number":3},{"id":"e97b4200b0cff27c87099ec0f8dd987a78a8ba0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-11-10 10:38:34.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b68157b026f6c8f1c15785515b3d7db85e6177a9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-10 10:50:15.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"bb109796d3178ad90d1658f5b21cd64da6161e51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 11:58:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e59f4d6c01b242c5a372ce7d0fd9e751\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bafae81ae1646f79c02fea2e3afe15f : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1f812a2d0714a2f803b853eeeae37b6 : SUCCESS in 7m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac255f7f260f46ce852343f875ef9c41 : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7665068dfd97411592be4fd418900ba8 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30dea25fa3674456b1e5f934aa87789f : SUCCESS in 9m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fcacfefae7f8484081635b8bce137e8d : FAILURE in 3m 45s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d864a428bcf546cdbd3475b313beeb7b : SUCCESS in 7m 32s\n- neutron-tempest-plugin-bgpvpn-bagpipe-wallaby https://zuul.opendev.org/t/openstack/build/7ae0abc6972444d2820b9563aaf251d6 : SUCCESS in 1h 06m 41s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/bc07f377df8249b7bf78300e12556b66 : FAILURE in 9m 08s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/c33d6a14d37642868d48372be064bc43 : SUCCESS in 1h 16m 58s","accounts_in_message":[],"_revision_number":4},{"id":"9e7717b09b61256014aebab115297c575520a646","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-11-10 13:04:21.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"03f85fd9b751de3a74588f9900680cd5bd8235c1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-10 13:34:58.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"74b89edc018c3732720cc0817b3e67ff7159022e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 13:35:35.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1c6f783e1e5d30c4fa12418e5bfc2f542ca38570","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 15:12:49.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8ec0a58dfb946368d40874c5a4a92e4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ea1841525714d708527bd6370f4fd7f : SUCCESS in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0e7af7108c54cf7bed1d335524d8acd : SUCCESS in 7m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c21f7885e5804cacbc9bbb643a5efbaa : SUCCESS in 7m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b52caf96c89b45b58ac0adfd23ec8167 : SUCCESS in 8m 59s\n- neutron-tempest-plugin-bgpvpn-bagpipe-wallaby https://zuul.opendev.org/t/openstack/build/7255407a4a4049b998f4685b5ce2d157 : SUCCESS in 1h 22m 40s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/1745a7ad86fc4cec8c0572282ec97361 : SUCCESS in 1h 32m 44s","accounts_in_message":[],"_revision_number":4},{"id":"4c7cb3e3c6906e8ac7997810e3dc16f140143d55","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 15:12:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"e927a8ce688abe4c7088ad93ab45503ef6240c29","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 15:14:45.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/940040cbe7ff42bfb422b89a314f4080\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69eb812741ab455b8f22d2449d5ee153 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"097de21d696c031972e6f05c8d1bbd3caff320ad","revisions":{"1f44d064188d222b54cc207125cfab89af3148eb":{"kind":"REWORK","_number":1,"created":"2022-10-24 16:13:48.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/02/862502/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/02/862502/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/02/862502/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/02/862502/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/02/862502/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/02/862502/1"}}},"commit":{"parents":[{"commit":"35b1177471c2ccd1acae1d951d52da2f1f756a5b","subject":"Merge \"Switch tempest jobs for stable/wallaby branch to wallaby jobs\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/35b1177471c2ccd1acae1d951d52da2f1f756a5b"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-10-13 18:23:10.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-10-24 16:12:26.000000000","tz":120},"subject":"Add required projects where necessary","message":"Add required projects where necessary\n\nMost of the jobs needs required-projects to be added, especially in\nstable branches, where the lack of them causes gate failures.\n\nSince the templates openstack-python3-jobs-neutron and\nperiodic-stable-jobs-neutron don\u0027t contain all the necessary projects\nfor bagpipe we need to add them one by one.\n\nChange-Id: Iceb92b157d5e875090a0dedec37c398716b89495\n(cherry picked from commit 7a02ee3188a4cadba76d75ea03400b35a33c2e5f)\n(cherry picked from commit 1b474c3333b25970431ade31d90a38af3f47fb7a)\n(cherry picked from commit d9bd4c660ee10c8de39d4b83090aefb79933fc00)\n(cherry picked from commit 39546f061193de8d46d6ca6141122f83bb18621a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/1f44d064188d222b54cc207125cfab89af3148eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/1f44d064188d222b54cc207125cfab89af3148eb"}]},"branch":"refs/heads/stable/wallaby"},"abed7961d3c89e0f8c26d4bceed4fab3a58f73d7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-10-28 15:45:18.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/02/862502/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/02/862502/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/02/862502/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/02/862502/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/02/862502/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/02/862502/2"}}},"commit":{"parents":[{"commit":"ad2729c1f8a4021b5c054e53b71c3fbaa7129333","subject":"Follow pyroute2 changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/ad2729c1f8a4021b5c054e53b71c3fbaa7129333"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-10-13 18:23:10.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2022-10-28 15:45:18.000000000","tz":0},"subject":"Add required projects where necessary","message":"Add required projects where necessary\n\nMost of the jobs needs required-projects to be added, especially in\nstable branches, where the lack of them causes gate failures.\n\nSince the templates openstack-python3-jobs-neutron and\nperiodic-stable-jobs-neutron don\u0027t contain all the necessary projects\nfor bagpipe we need to add them one by one.\n\nChange-Id: Iceb92b157d5e875090a0dedec37c398716b89495\n(cherry picked from commit 7a02ee3188a4cadba76d75ea03400b35a33c2e5f)\n(cherry picked from commit 1b474c3333b25970431ade31d90a38af3f47fb7a)\n(cherry picked from commit d9bd4c660ee10c8de39d4b83090aefb79933fc00)\n(cherry picked from commit 39546f061193de8d46d6ca6141122f83bb18621a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/abed7961d3c89e0f8c26d4bceed4fab3a58f73d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/abed7961d3c89e0f8c26d4bceed4fab3a58f73d7"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"},"a9b85a61a3aa4a32865eb79e3b2d1993f8d696d2":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-11-09 17:24:05.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/02/862502/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/02/862502/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/02/862502/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/02/862502/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/02/862502/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/02/862502/3"}}},"commit":{"parents":[{"commit":"ad2729c1f8a4021b5c054e53b71c3fbaa7129333","subject":"Follow pyroute2 changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/ad2729c1f8a4021b5c054e53b71c3fbaa7129333"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-10-13 18:23:10.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2022-11-09 17:24:05.000000000","tz":0},"subject":"Add required projects where necessary","message":"Add required projects where necessary\n\nMost of the jobs needs required-projects to be added, especially in\nstable branches, where the lack of them causes gate failures.\n\nSince the templates openstack-python3-jobs-neutron and\nperiodic-stable-jobs-neutron don\u0027t contain all the necessary projects\nfor bagpipe we need to add them one by one.\n\nChanges:\n  .zuul.yaml\n\nChange-Id: Iceb92b157d5e875090a0dedec37c398716b89495\n(cherry picked from commit 7a02ee3188a4cadba76d75ea03400b35a33c2e5f)\n(cherry picked from commit 1b474c3333b25970431ade31d90a38af3f47fb7a)\n(cherry picked from commit d9bd4c660ee10c8de39d4b83090aefb79933fc00)\n(cherry picked from commit 39546f061193de8d46d6ca6141122f83bb18621a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a9b85a61a3aa4a32865eb79e3b2d1993f8d696d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a9b85a61a3aa4a32865eb79e3b2d1993f8d696d2"}]},"branch":"refs/heads/stable/wallaby","description":"Edit commit message"},"097de21d696c031972e6f05c8d1bbd3caff320ad":{"kind":"REWORK","_number":4,"created":"2022-11-10 10:38:34.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/02/862502/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/02/862502/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/02/862502/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/02/862502/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/02/862502/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/02/862502/4"}}},"commit":{"parents":[{"commit":"35b1177471c2ccd1acae1d951d52da2f1f756a5b","subject":"Merge \"Switch tempest jobs for stable/wallaby branch to wallaby jobs\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/35b1177471c2ccd1acae1d951d52da2f1f756a5b"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-07-08 14:04:55.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-11-10 10:35:39.000000000","tz":60},"subject":"Add required project and follow pyroute2 changes","message":"Add required project and follow pyroute2 changes\n\nThis is a combination of 2 commits to unblock the gate.\n\n1. Follow pyroute2 changes\n\npyroute2 recently changed its module structure (see [1]) and deprecated\nIPDB (see [2]), this patch tries to address these.\n\n[1]: https://github.com/svinota/pyroute2/discussions/786\n[2]: https://github.com/svinota/pyroute2/discussions/788\n\n(cherry picked from commit b01bc67ab803d89754f2da016761301dad353501)\n\n2. Add required projects where necessary\n\nMost of the jobs needs required-projects to be added, especially in\nstable branches, where the lack of them causes gate failures.\n\nSince the templates openstack-python3-jobs-neutron and\nperiodic-stable-jobs-neutron don\u0027t contain all the necessary projects\nfor bagpipe we need to add them one by one.\n\nChanges:\n  .zuul.yaml\n\nChange-Id: Iceb92b157d5e875090a0dedec37c398716b89495\n(cherry picked from commit 7a02ee3188a4cadba76d75ea03400b35a33c2e5f)\n(cherry picked from commit 1b474c3333b25970431ade31d90a38af3f47fb7a)\n(cherry picked from commit d9bd4c660ee10c8de39d4b83090aefb79933fc00)\n(cherry picked from commit 39546f061193de8d46d6ca6141122f83bb18621a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/097de21d696c031972e6f05c8d1bbd3caff320ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/097de21d696c031972e6f05c8d1bbd3caff320ad"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[]}
