)]}'
{"id":"openstack%2Fnetworking-bagpipe~872632","triplet_id":"openstack%2Fnetworking-bagpipe~master~I655e301097ce154fb7f823680b998239bb7b4dde","project":"openstack/networking-bagpipe","branch":"master","topic":"dnm","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-02-03 15:14:33.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I655e301097ce154fb7f823680b998239bb7b4dde","subject":"DNM: check master","status":"ABANDONED","created":"2023-02-03 09:13:05.000000000","updated":"2023-02-03 15:14:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5784559e6bb5018eabf7407d0537b3141fb91274","_number":872632,"virtual_id_number":872632,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-02-03 10:37:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-03 10:37:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ed4a9ae22fafea2fd70d6901cf3ae7204a7262b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-02-03 09:13:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd9e1e4253a2e982f838dfa56ec35cab60b41648","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 10:37:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e822ffd0ac1f49a9b8590f117775f2a5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bde73ff1d9b742e0bb1f32bca87e415d : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7061c1bef152431ba71005685b9c188c : FAILURE in 7m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb0c69cc58c54cb991af9877b2ec1a2d : FAILURE in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78333e0a89f2461f87d3cd6c75acf0ba : SUCCESS in 9m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aaea8d5354e54ac68c12ff267115445d : FAILURE in 8m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e31f2d9487144289374b073d2e548c7 : FAILURE in 7m 27s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/25129b8a730f424bb21597a7b40e9dbd : SUCCESS in 1h 13m 19s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/53ba91c6efe64d5aa6067b3e347c97c2 : FAILURE in 8m 20s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/37eaf817daaf48b2b4288aa076d6f7b3 : SUCCESS in 1h 20m 29s","accounts_in_message":[],"_revision_number":1},{"id":"5784559e6bb5018eabf7407d0537b3141fb91274","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-02-03 15:14:33.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"84fa83c3ffa650c2c35cb86d181b9a06407f9ffb","revisions":{"84fa83c3ffa650c2c35cb86d181b9a06407f9ffb":{"kind":"REWORK","_number":1,"created":"2023-02-03 09:13:05.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/32/872632/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/32/872632/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/32/872632/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/32/872632/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/32/872632/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/32/872632/1"}}},"commit":{"parents":[{"commit":"0d094ba1f24393e03446b22d9cb5550a94287455","subject":"Tox4: add allowlist_externals where necessary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0d094ba1f24393e03446b22d9cb5550a94287455"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-02-03 09:12:39.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-02-03 09:12:39.000000000","tz":60},"subject":"DNM: check master","message":"DNM: check master\n\nChange-Id: I655e301097ce154fb7f823680b998239bb7b4dde\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/84fa83c3ffa650c2c35cb86d181b9a06407f9ffb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/84fa83c3ffa650c2c35cb86d181b9a06407f9ffb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
