)]}'
{"id":"openstack%2Fnetworking-bagpipe~883064","triplet_id":"openstack%2Fnetworking-bagpipe~master~I2f643e539df39a36e81570257ab9425131d02b9d","project":"openstack/networking-bagpipe","branch":"master","topic":"bug/2019097","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-06-05 14:39:38.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-06-08 19:30:45.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-06-08 16:50:33.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-06-08 19:30:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2f643e539df39a36e81570257ab9425131d02b9d","subject":"Fix issues due to rcent RBAC changes and removal of neutron.debug","status":"MERGED","created":"2023-05-12 12:33:43.000000000","updated":"2023-06-08 19:31:38.000000000","submitted":"2023-06-08 19:30:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"883064-bug/2019097","meta_rev_id":"3bf6f8bb0a582a23b494397ca38017c3bbfbb11e","_number":883064,"virtual_id_number":883064,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-08 19:30:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-06-08 13:00:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-06-08 13:00:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-06-05 14:39:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-12 14:16:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-16 10:04:44.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-05-24 18:03:42.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-05-31 13:24:16.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"c6ac130c7aac3def90479cb6d610c6ed9c51d528","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-05-12 12:33:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7881e66728f461cfe21f854c4c26147f78ae651d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-12 14:16:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84b5b46060d548dd9252b1597cd5dbde\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/874ce4c1ec174a6c9ef0cf7e375b324c : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f449f41ed11404691f6a1c802341858 : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bf1cd108da14fba9ba6fe9a4ab03f57 : SUCCESS in 7m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6b6fb3d081ad48de9a06ddaf815449c7 : SUCCESS in 7m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79aa63a9c47741eaacb11d2df8e49f05 : SUCCESS in 10m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d91bfaf8653341debfd02a8ba3dadd60 : SUCCESS in 2m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdbabdf2740c448db815e9e5f69c1ad8 : SUCCESS in 8m 04s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/6bc6e54c413a420c8e0792174afb5940 : SUCCESS in 1h 35m 37s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/0d2ff7108b624377982506b0d9dc4bfe : FAILURE in 21m 48s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/3dd351fd3f9542779d0a45037763432b : SUCCESS in 1h 22m 24s","accounts_in_message":[],"_revision_number":1},{"id":"05b705df494d3a478d6aaf19d75ca3c2e8634997","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-16 10:04:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d26251f9faafea46cfe1d3365c709861e5c1c0f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 10:05:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b49f135ff239a79f7f6a21186f1c94cf0e1bcd5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-16 10:16:56.000000000","message":"Patch Set 1: Code-Review-1 -Workflow\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"780a31c95c4da098f4e2e695fee83be74aa7dc42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 10:17:12.000000000","message":"Patch Set 1: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":1},{"id":"fa2f33f0564793ba2823afa37fbd19456e875512","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-05-16 11:47:22.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n","accounts_in_message":[],"_revision_number":2},{"id":"28ab15e6cc0d262db6427fb447c0feecb689d7bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-05-16 11:54:03.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"4291724273ea4ffcae6bc1ef18f79b291909bf5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 13:20:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2cc31d6d474246f7a3a594586e3e6a93\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d8a35153e244c8fa0ff300afe875d4e : FAILURE in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8da4cae7863f4cb9afab3fd7309fc70a : FAILURE in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/086ca0466f1a405c88f6b898abea6ac1 : FAILURE in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/08a60456dd694afc9de62c444fb17127 : FAILURE in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2539c384c56249d49bb0c3a8ec29cb17 : FAILURE in 5m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d789a2279c9d4102aac6d7f84ad270d0 : SUCCESS in 2m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af7c1ada92b54bc9aac96ac63a33fffd : FAILURE in 3m 28s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/079c59c6ddd140b98c9538d4d39b5f52 : FAILURE in 33m 37s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/55953226ca694e49ab520445d128f6da : FAILURE in 19m 56s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/4aa2b94ffdef4637929ecfdfa73d55d2 : SUCCESS in 1h 21m 03s","accounts_in_message":[],"_revision_number":3},{"id":"8a2bf2a9dca2e71f7973c41cb32b3c43ca20f4a0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-17 08:45:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"636fb15940b9a3a0ab58a85a7f62d499a3ffc4fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 10:53:58.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0278a6071951406e8a8bae3e2c574c21\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaaa906bf05248a3a251646371fc32f6 : FAILURE in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/680c484541804798bb605a6d9fdcf624 : FAILURE in 6m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99b0c35066284d8281da855b4811eb70 : FAILURE in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f5dab9c86a9d45098d5c40e2a3078524 : FAILURE in 7m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77ce6c16b30a47bc839e7dc4e0628611 : SUCCESS in 10m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1d074b5745354de0b9517abd54f2b819 : SUCCESS in 2m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9233049e58cd466882eae08234fe6751 : FAILURE in 6m 09s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/e18ef6120b444b2c89a871de7feb0d91 : TIMED_OUT in 2h 02m 59s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/8ec681eff2804c0ab8d13f9075515716 : FAILURE in 17m 45s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/d55d999aa3504a69ad60a4d78d11c79c : SUCCESS in 1h 24m 27s\n\nWarning:\n  Comments left for invalid file networking_bagpipe/agent/bgpvpn/agent_extension.py","accounts_in_message":[],"_revision_number":3},{"id":"3c039194f47eacbd30235dc796a03279b4bfb0a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-05-17 12:46:17.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"9a1db5114439dd258916b25503217afd58121d34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 14:27:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/02b4784f16174bdfb31b983077224a18\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb81bb06e24f49a097bcff884862c56e : FAILURE in 8m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca9fccee2f3140ca8af7760f873ec3af : SUCCESS in 7m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5fe16cba478e4d75ad21b44770b2bbbc : SUCCESS in 8m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/16ab5abfdd6d405292e8613f554f93b4 : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01fac458f1f4471189f2a909eb4a889e : SUCCESS in 11m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5cc0057de3ba49619c7c52ed25a37612 : SUCCESS in 3m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee5330ef5fab452e982db094d594e326 : SUCCESS in 7m 34s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/6854f6c2a6894ed589a6aac0183dda61 : SUCCESS in 1h 39m 40s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/0d53defdbe0b4a27aaafd60dd4ceab3c : FAILURE in 7m 58s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/a34fb7aef23f45ad8d7b425c017b3b51 : SUCCESS in 1h 08m 37s","accounts_in_message":[],"_revision_number":4},{"id":"6446820958b72addec07b2fd62f35c2b324b8a68","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-18 10:38:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f4b06c1921f20388ba3d196cef674fe4c0ec75a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 12:21:57.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b206eba64f3c400d969b19ad595fa7b1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f926b8ed4e14f12ab6c70fee87bacb6 : FAILURE in 8m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3872aea9926d48d090aa4418be263d15 : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/053e018558574cb6bd206a9ceda5883a : SUCCESS in 7m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e67faa9e4af74115a971c1cbaafeae6f : SUCCESS in 8m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b9e09daf66c43818b69d2fd3bbe5f67 : SUCCESS in 11m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5aa9340845e24194ad4234378ad7d2be : SUCCESS in 2m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/654efe4784be418caf7e2029b10b5369 : SUCCESS in 7m 34s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/34192829b44f4aee873fc813f12a4505 : SUCCESS in 1h 35m 56s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/0604d3c79a4b44698a6de26e7ca0fcfb : FAILURE in 17m 54s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/a9a4bbec05454d3ea54991291e384817 : SUCCESS in 1h 30m 13s","accounts_in_message":[],"_revision_number":4},{"id":"43fed00c4702a18d9bde867e940b67016157af7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-05-22 10:36:42.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"aa55d5ff19f97818c1bbabb7ddf0d47f08c0bfeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 12:19:25.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/625813b3e09c493e894ffcffff5e74d0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4806c47ac954565abef20d156ad5b54 : FAILURE in 8m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38ccee7885c642e4be55bbe61f5be56a : SUCCESS in 9m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56583112dd8d4697875458ed6bea8766 : SUCCESS in 9m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86ab905320664861a0713e22478447cb : SUCCESS in 9m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/761210e7ed1845f7b3ab15449109af2f : SUCCESS in 10m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/89be1ef0e3b649e6bd0bbec67bdeb92d : SUCCESS in 2m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/126a743c7b2c4a29ae4636368c9896c9 : SUCCESS in 10m 09s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/0d2b4bf09b40493d9eaa44c02ade14ba : SUCCESS in 1h 37m 04s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/2e9e6d91582f4223883c9636b5918424 : FAILURE in 9m 00s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/8ef17f4d3ede409f8db6f2a763731cc5 : SUCCESS in 1h 17m 11s","accounts_in_message":[],"_revision_number":5},{"id":"ecd5f26c62effb90c4089e9b24d3320071478e1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-05-22 13:00:05.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"62f567f86feb3bcd2f55309370094864f3f792f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 14:31:25.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/81c3c50b8ac64efda44efe4fbe313c91\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e37e055ac1b24c619f573f266dadf08b : SUCCESS in 9m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ffcdc7e4fb94878827eea78ff266ac5 : SUCCESS in 7m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7cea707d95124bf58f94eaac004d484a : SUCCESS in 7m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6cfdecfadefd4910a3d8b3a1313d35a7 : SUCCESS in 8m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5502b7381b3844fe9b6b6246b10d087d : SUCCESS in 9m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84ba92e1f4f741b28a200f32150af342 : SUCCESS in 4m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5ee802ee7c9d4baeae28fc5a82852ed1 : SUCCESS in 2m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51b5bc5b592546ca838b7e62f62db707 : SUCCESS in 9m 14s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/6442bd60b41b4506afd7f29612a0572a : FAILURE in 52m 54s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/f873fb07c3514eeb9e3a3c6a1d292fd2 : FAILURE in 17m 29s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/fe7cfb89f69142b5837bbc6e6d3f9464 : SUCCESS in 1h 20m 22s","accounts_in_message":[],"_revision_number":6},{"id":"3ae4f035e98eb13912594cf34cb3d2bdf45452f2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-05-24 18:03:42.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d9434fa42e64e03f7374cc612a188c856a0e72ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-05-31 10:43:57.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"ce0e191606097aa971151cec98d958f62b2928ff","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-05-31 10:46:47.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"658f9c439ec0f1195dffba481d6a1343bb0448d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-05-31 10:55:02.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"35b618c9ac0f3409fed9bbda37b9dcdaa7e25727","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 12:43:52.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f030e2048c2843d7b3670e0649e9ce70\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d143f50653da46928385589879b198c4 : SUCCESS in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49fe2e0be0c449a481e0ea70270e3788 : SUCCESS in 7m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb4d2866dbeb4d4b8995fb61eafd196b : SUCCESS in 9m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f8e4b48d0d343068d85737e3c2fe5e2 : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b4ea4d603bd4f7b98b3e6695a1b0031 : SUCCESS in 12m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ec2fb050ffe40a19dd890dd676b0c60 : SUCCESS in 4m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7927eceafbe5411c88328be9dfa9c048 : SUCCESS in 3m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c09000a677944ff931e7721bdd8f9c0 : SUCCESS in 7m 33s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/2b51e42ca3d948efb8f06eb6abe0a8c6 : SUCCESS in 1h 38m 46s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/686b77ec7ed64cfc8a47bc534adfbcd4 : FAILURE in 9m 18s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/a9b7b414d4944dc6815e42d151ede6dd : SUCCESS in 1h 20m 44s","accounts_in_message":[],"_revision_number":8},{"id":"a97425f0ba824204141b6a7cbee452c0965c09e1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-31 13:21:13.000000000","message":"Patch Set 8: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":8},{"id":"8d3f66a73f6647678719f2fac7c756d02ab92d1e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-05-31 13:24:16.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"a0ca26e4b81a9a4b9961a9b4a245f8f3ac419b43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 13:24:28.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/61726b2a311043f993d383e9a1427dea\n\n\nWarning:\n  Change 883057 in project openstack/networking-bgpvpn does not share a change queue with 883064 in project openstack/networking-bagpipe","accounts_in_message":[],"_revision_number":8},{"id":"dfc8d442fc62ed6de8baea42e9012ebc657c832d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-01 10:22:19.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3333dd5bbae2d6c3de81119a4965d37f988580f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 12:14:46.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fb941c3235a49e0a492cfe304e036ac\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77ca92547a2243e19b011ac0d3e499a5 : SUCCESS in 9m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03c1de192dc745278d2af7b7f81f2b98 : SUCCESS in 8m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c78b7b76d654d44a22ab7f022f4618b : SUCCESS in 9m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/919916648aae4af3ae4d737018da4e0b : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c48f936c28d4a2c905e7881ce2e4774 : SUCCESS in 11m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0e0fa39561b4983af0cec96f37270c1 : SUCCESS in 4m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/be8272f46b4c49efa2bc089c160e8b72 : SUCCESS in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d50c9d851264139a916f77d3a38d33e : SUCCESS in 9m 01s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/3799dec43c1e44a08eafad61d3c77034 : SUCCESS in 1h 42m 14s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/faa8641e7ad14a37a22ff8fa08652ecb : FAILURE in 9m 04s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/fd99cba1c4654f54928cf61b4c275171 : SUCCESS in 1h 40m 34s","accounts_in_message":[],"_revision_number":8},{"id":"09afb3c582fd9f14abc805f86cc02c2349cfaa56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 12:14:59.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/04061024f63d49e8ab86574a16df6f7b\n\n\nWarning:\n  Change 883057 in project openstack/networking-bgpvpn does not share a change queue with 883064 in project openstack/networking-bagpipe","accounts_in_message":[],"_revision_number":8},{"id":"4b649cdaa3d1ce89627c3f78aee5fe3a5ae8bbb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-06-05 08:34:10.000000000","message":"Patch Set 9: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":9},{"id":"2dd422d1216c4da79d8cb2629c5d8a38023bbbdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 08:34:23.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a94862fcdcf4bf0a378a78f1a2ba51b\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":9},{"id":"f91b2aa5aafc0ca1362bef9eef2617199e425526","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-05 14:39:38.000000000","message":"Patch Set 9: Code-Review-1 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"50b22e61aa8a0bd478bfec46c416eaf8066c4a49","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-06-06 12:11:10.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"e45a3e529bf50e8bc001245e5ca9b71cd6ed55f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 13:32:43.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc638908ca6d474f9b0ce7d8917d19ff\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53e53b55c7bc4f69ab5ad74b105e2960 : SUCCESS in 7m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36f543bd7197480db38c872634eb46cc : SUCCESS in 7m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a43a14434c184704b4537261cabb1620 : SUCCESS in 8m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c23900ce619346849c95c3f1c559714d : SUCCESS in 8m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6ae765a4b424dce9a60dca3bb8ca939 : SUCCESS in 12m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3483d6bb2064d1e9f0924962f9f1178 : SUCCESS in 4m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bd708a6b838d466c853f51051b6ad385 : SUCCESS in 2m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0245b084e0c445619e28012a702e9a2a : SUCCESS in 9m 23s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/eefe8f51d44b492982393ede84cfa998 : SUCCESS in 1h 19m 21s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/dbde047c593444799c9206d35467cef4 : FAILURE in 10m 58s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/3018e1a3205b4a7daf39fefd0056eea1 : SUCCESS in 1h 00m 17s","accounts_in_message":[],"_revision_number":9},{"id":"6fb53947c60fbcf35739df922ed41d920461c2dd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-08 13:00:58.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"6f27239b001256b3461344f7c6da4decdf85d47e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-08 13:01:09.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c45de00ef51c4c1caa7bc361b6788179\n\n\nWarning:\n  Change 883057 in project openstack/networking-bgpvpn does not share a change queue with 883064 in project openstack/networking-bagpipe","accounts_in_message":[],"_revision_number":9},{"id":"e8506f1523765508d9518ce72d1eca222663bdb1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-06-08 16:50:33.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"e72b37dc1187fd2edec850bb16bf91fa61e2a4a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-08 18:18:54.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4412577f5bb4a01be55e54034dd263d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c634b4bd54e4ac1873318b59d1e7f11 : SUCCESS in 8m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a544abdae614911916abc2cbe345771 : SUCCESS in 8m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74409e228e1e40f6b00750874ee6c73d : SUCCESS in 8m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/43fa706e6ac944f7be0b9eba31e11c39 : SUCCESS in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc50bc0f286d4d3fb6e2b70b7f6cdbfc : SUCCESS in 11m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afa9a28fbe1d45849fa23e41d0ad3fe0 : SUCCESS in 3m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8b0c801221174e66b225811f60b3b4bb : SUCCESS in 2m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81fbff98b6a74f1f825d4a38d6717963 : SUCCESS in 7m 26s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/14d8ddea4a4e41b6aaca7814120408ac : SUCCESS in 1h 26m 23s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/c1ccb27ea7cc44afa118bf3ef4986f74 : FAILURE in 9m 07s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/51d9a59321534ea391ef12efd4082fd9 : SUCCESS in 1h 07m 05s","accounts_in_message":[],"_revision_number":9},{"id":"d345111098f51834b8b982de2207f78226dff6fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-08 18:19:24.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"88c81fcdeef44d1ffdb1eb0f8e956eae9bb60d6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-08 19:30:45.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2da6d1d76fa54213bf47552a60779e50\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ffee457749a46d090c21e546387e55e : SUCCESS in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fb7b01b6ca24e2da4a4814be53bccfa : SUCCESS in 6m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c5ecd50752d4eee841a8ec0fde898ab : SUCCESS in 7m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb05aa7cb6724bd3b543654ac4a90c11 : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd56e416d5f2442f90a3f7d1dda4988d : SUCCESS in 9m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20b76cf23ec14728ae827fc048fcc408 : SUCCESS in 4m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fcf7a3390b0b47e6982fb3c7502992c1 : SUCCESS in 3m 02s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/0030ff9a9e4a42b38bbd85fb61158ee9 : SUCCESS in 1h 10m 46s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/01c50b932feb465eb0152d92871eb279 : SUCCESS in 1h 05m 04s","accounts_in_message":[],"_revision_number":9},{"id":"7bdfa715a815ab3f6b0b60bd89f91b9b87e958f1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-08 19:30:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"3bf6f8bb0a582a23b494397ca38017c3bbfbb11e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-08 19:31:38.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85dbd55503ee4c099a5f9580ffa104f6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5c62a4b64ec497497a19ab1a0622b8a : SUCCESS in 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/865469ff64014a3688eb8dbce6f0745d : SUCCESS in 35s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"23a3f6026e5c0a2026804e71b17c458751c7844e","revisions":{"f8164fe4051e91511f4d929fd68ffa457e28c8c4":{"kind":"REWORK","_number":1,"created":"2023-05-12 12:33:43.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/64/883064/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/64/883064/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/1"}}},"commit":{"parents":[{"commit":"59df23dd75b8e40a64b3fdfde44976691c8a8621","subject":"Merge \"CI: Add periodic weekly job with sqlalchemy main\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/59df23dd75b8e40a64b3fdfde44976691c8a8621"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-05-12 12:32:58.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-05-12 12:32:58.000000000","tz":120},"subject":"Fix issues due to rcent RBAC changes","message":"Fix issues due to rcent RBAC changes\n\nneutron recently merged a commit enabling secure rbacs by default [1]\nbreaking several unit tests of networking-bagpipe.\nThis patch changes the necessary test calls to be admin and\nrequires neutron \u003e\u003d 23.0.0.0b1.\n\n[1]: https://opendev.org/openstack/neutron/commit/670cc383e0345063c0bf9022f8829d5817457e00\n\nChange-Id: I2f643e539df39a36e81570257ab9425131d02b9d\nRelated-Bug: 2019097\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f8164fe4051e91511f4d929fd68ffa457e28c8c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f8164fe4051e91511f4d929fd68ffa457e28c8c4"}]},"branch":"refs/heads/master"},"e867d1eb83acc412048d34e77f457e10c5a79938":{"kind":"REWORK","_number":2,"created":"2023-05-16 11:47:22.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/64/883064/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/64/883064/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/2"}}},"commit":{"parents":[{"commit":"59df23dd75b8e40a64b3fdfde44976691c8a8621","subject":"Merge \"CI: Add periodic weekly job with sqlalchemy main\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/59df23dd75b8e40a64b3fdfde44976691c8a8621"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-05-12 12:32:58.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2023-05-16 11:47:22.000000000","tz":0},"subject":"Fix issues due to rcent RBAC changes","message":"Fix issues due to rcent RBAC changes\n\nneutron recently merged a commit enabling secure rbacs by default [1]\nbreaking several unit tests of networking-bagpipe.\nThis patch changes the necessary test calls to be admin and\nrequires neutron \u003e\u003d 23.0.0.0b1.\n\n[1]: https://opendev.org/openstack/neutron/commit/670cc383e0345063c0bf9022f8829d5817457e00\n\nChange-Id: I2f643e539df39a36e81570257ab9425131d02b9d\nRelated-Bug: 2019097\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e867d1eb83acc412048d34e77f457e10c5a79938"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e867d1eb83acc412048d34e77f457e10c5a79938"}]},"branch":"refs/heads/master"},"9255f83a13833f8fc34fb901c4a96058096e84fd":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-05-16 11:54:03.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/64/883064/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/64/883064/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/3"}}},"commit":{"parents":[{"commit":"59df23dd75b8e40a64b3fdfde44976691c8a8621","subject":"Merge \"CI: Add periodic weekly job with sqlalchemy main\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/59df23dd75b8e40a64b3fdfde44976691c8a8621"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-05-12 12:32:58.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2023-05-16 11:54:02.000000000","tz":0},"subject":"Fix issues due to rcent RBAC changes","message":"Fix issues due to rcent RBAC changes\n\nneutron recently merged a commit enabling secure rbacs by default [1]\nbreaking several unit tests of networking-bagpipe.\nThis patch changes the necessary test calls to be admin and\nrequires neutron \u003e\u003d 23.0.0.0b2.\n\n[1]: https://opendev.org/openstack/neutron/commit/670cc383e0345063c0bf9022f8829d5817457e00\n\nChange-Id: I2f643e539df39a36e81570257ab9425131d02b9d\nRelated-Bug: 2019097\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/9255f83a13833f8fc34fb901c4a96058096e84fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/9255f83a13833f8fc34fb901c4a96058096e84fd"}]},"branch":"refs/heads/master","description":"Edit commit message"},"0a1692486981061a8373cac9ebcc7a7e4d6feb30":{"kind":"REWORK","_number":4,"created":"2023-05-17 12:46:17.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/64/883064/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/64/883064/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/4"}}},"commit":{"parents":[{"commit":"59df23dd75b8e40a64b3fdfde44976691c8a8621","subject":"Merge \"CI: Add periodic weekly job with sqlalchemy main\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/59df23dd75b8e40a64b3fdfde44976691c8a8621"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-05-12 12:32:58.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-05-17 12:45:54.000000000","tz":120},"subject":"Fix issues due to rcent RBAC changes and removal of neutron.debug","message":"Fix issues due to rcent RBAC changes and removal of neutron.debug\n\nneutron recently merged a commit enabling secure rbacs by default [1]\nbreaking several unit tests of networking-bagpipe.\nThis patch changes the necessary test calls to be admin and\nrequires neutron \u003e\u003d 23.0.0.0b2.\n[2] remove neutron.debug, remove the dependency to it in\nagent_extension.\n\n[1]: https://opendev.org/openstack/neutron/commit/670cc383e0345063c0bf9022f8829d5817457e00\n[2]: https://review.opendev.org/c/openstack/neutron/+/883081\n\nChange-Id: I2f643e539df39a36e81570257ab9425131d02b9d\nRelated-Bug: 2019097\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0a1692486981061a8373cac9ebcc7a7e4d6feb30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0a1692486981061a8373cac9ebcc7a7e4d6feb30"}]},"branch":"refs/heads/master"},"c85c6994ed7eb972f1003f317615b2846e007e87":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2023-05-22 10:36:42.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/64/883064/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/64/883064/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/5"}}},"commit":{"parents":[{"commit":"59df23dd75b8e40a64b3fdfde44976691c8a8621","subject":"Merge \"CI: Add periodic weekly job with sqlalchemy main\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/59df23dd75b8e40a64b3fdfde44976691c8a8621"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-05-12 12:32:58.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2023-05-22 10:36:42.000000000","tz":0},"subject":"Fix issues due to rcent RBAC changes and removal of neutron.debug","message":"Fix issues due to rcent RBAC changes and removal of neutron.debug\n\nneutron recently merged a commit enabling secure rbacs by default [1]\nbreaking several unit tests of networking-bagpipe.\nThis patch changes the necessary test calls to be admin and\nrequires neutron \u003e\u003d 23.0.0.0b2.\n[2] remove neutron.debug, remove the dependency to it in\nagent_extension.\n\n[1]: https://opendev.org/openstack/neutron/commit/670cc383e0345063c0bf9022f8829d5817457e00\n[2]: https://review.opendev.org/c/openstack/neutron/+/883081\n\nDepends-On: https://review.opendev.org/c/openstack/networking-bgpvpn/+/883057\nChange-Id: I2f643e539df39a36e81570257ab9425131d02b9d\nRelated-Bug: 2019097\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c85c6994ed7eb972f1003f317615b2846e007e87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c85c6994ed7eb972f1003f317615b2846e007e87"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7634d891e4cac3a5b12763d12496ac74db713c45":{"kind":"REWORK","_number":6,"created":"2023-05-22 13:00:05.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/64/883064/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/64/883064/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/6"}}},"commit":{"parents":[{"commit":"59df23dd75b8e40a64b3fdfde44976691c8a8621","subject":"Merge \"CI: Add periodic weekly job with sqlalchemy main\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/59df23dd75b8e40a64b3fdfde44976691c8a8621"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-05-12 12:32:58.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-05-22 12:59:16.000000000","tz":120},"subject":"Fix issues due to rcent RBAC changes and removal of neutron.debug","message":"Fix issues due to rcent RBAC changes and removal of neutron.debug\n\nneutron recently merged a commit enabling secure rbacs by default [1]\nbreaking several unit tests of networking-bagpipe.\nThis patch changes the necessary test calls to be admin and\nrequires neutron \u003e\u003d 23.0.0.0b2.\n[2] remove neutron.debug, remove the dependency to it in\nagent_extension.\nRemove skipdist\u003dTrue from tox.ini to avoid bagpipe installaion\nfrom package.\n\n[1]: https://opendev.org/openstack/neutron/commit/670cc383e0345063c0bf9022f8829d5817457e00\n[2]: https://review.opendev.org/c/openstack/neutron/+/883081\n\nDepends-On: https://review.opendev.org/c/openstack/networking-bgpvpn/+/883057\nChange-Id: I2f643e539df39a36e81570257ab9425131d02b9d\nRelated-Bug: 2019097\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7634d891e4cac3a5b12763d12496ac74db713c45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7634d891e4cac3a5b12763d12496ac74db713c45"}]},"branch":"refs/heads/master"},"57c0d84b480bd930c01ae048c4b568efe669cd35":{"kind":"REWORK","_number":7,"created":"2023-05-31 10:43:57.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/64/883064/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/64/883064/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/7"}}},"commit":{"parents":[{"commit":"59df23dd75b8e40a64b3fdfde44976691c8a8621","subject":"Merge \"CI: Add periodic weekly job with sqlalchemy main\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/59df23dd75b8e40a64b3fdfde44976691c8a8621"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-05-12 12:32:58.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-05-31 10:43:15.000000000","tz":120},"subject":"Fix issues due to rcent RBAC changes and removal of neutron.debug","message":"Fix issues due to rcent RBAC changes and removal of neutron.debug\n\nneutron recently merged a commit enabling secure rbacs by default [1]\nbreaking several unit tests of networking-bagpipe.\nThis patch changes the necessary test calls to be admin and\nrequires neutron \u003e\u003d 23.0.0.0b2.\n[2] removed neutron.debug, remove the dependency to it in\nagent_extension.\nRemove skipdist\u003dTrue from tox.ini to avoid bagpipe installaion\nfrom package.\n\n[1]: https://opendev.org/openstack/neutron/commit/670cc383e0345063c0bf9022f8829d5817457e00\n[2]: https://review.opendev.org/c/openstack/neutron/+/883081\n\nDepends-On: https://review.opendev.org/c/openstack/networking-bgpvpn/+/883057\nChange-Id: I2f643e539df39a36e81570257ab9425131d02b9d\nRelated-Bug: 2019097\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/57c0d84b480bd930c01ae048c4b568efe669cd35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/57c0d84b480bd930c01ae048c4b568efe669cd35"}]},"branch":"refs/heads/master"},"61f56ac03168bbe83b30d85c039d5604c5e9649b":{"kind":"REWORK","_number":8,"created":"2023-05-31 10:55:02.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/64/883064/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/64/883064/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/8"}}},"commit":{"parents":[{"commit":"59df23dd75b8e40a64b3fdfde44976691c8a8621","subject":"Merge \"CI: Add periodic weekly job with sqlalchemy main\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/59df23dd75b8e40a64b3fdfde44976691c8a8621"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-05-12 12:32:58.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-05-31 10:54:29.000000000","tz":120},"subject":"Fix issues due to rcent RBAC changes and removal of neutron.debug","message":"Fix issues due to rcent RBAC changes and removal of neutron.debug\n\nneutron recently merged a commit enabling secure rbacs by default [1]\nbreaking several unit tests of networking-bagpipe.\nThis patch changes the necessary test calls to be admin and\nrequires neutron \u003e\u003d 23.0.0.0b2.\n[2] removed neutron.debug, remove the dependency to it in\nagent_extension.\nRemove skipdist\u003dTrue from tox.ini to avoid bagpipe installaion\nfrom package.\n\n[1]: https://opendev.org/openstack/neutron/commit/670cc383e0345063c0bf9022f8829d5817457e00\n[2]: https://review.opendev.org/c/openstack/neutron/+/883081\n\nDepends-On: https://review.opendev.org/c/openstack/networking-bgpvpn/+/883057\nChange-Id: I2f643e539df39a36e81570257ab9425131d02b9d\nRelated-Bug: 2019097\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/61f56ac03168bbe83b30d85c039d5604c5e9649b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/61f56ac03168bbe83b30d85c039d5604c5e9649b"}]},"branch":"refs/heads/master"},"23a3f6026e5c0a2026804e71b17c458751c7844e":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2023-06-05 08:34:10.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/64/883064/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/64/883064/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/64/883064/9"}}},"commit":{"parents":[{"commit":"59df23dd75b8e40a64b3fdfde44976691c8a8621","subject":"Merge \"CI: Add periodic weekly job with sqlalchemy main\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/59df23dd75b8e40a64b3fdfde44976691c8a8621"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-05-12 12:32:58.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2023-06-05 08:34:10.000000000","tz":0},"subject":"Fix issues due to rcent RBAC changes and removal of neutron.debug","message":"Fix issues due to rcent RBAC changes and removal of neutron.debug\n\nneutron recently merged a commit enabling secure rbacs by default [1]\nbreaking several unit tests of networking-bagpipe.\nThis patch changes the necessary test calls to be admin and\nrequires neutron \u003e\u003d 23.0.0.0b2.\n[2] removed neutron.debug, remove the dependency to it in\nagent_extension.\nRemove skipdist\u003dTrue from tox.ini to avoid bagpipe installaion\nfrom package.\n\n[1]: https://opendev.org/openstack/neutron/commit/670cc383e0345063c0bf9022f8829d5817457e00\n[2]: https://review.opendev.org/c/openstack/neutron/+/883081\n\nChange-Id: I2f643e539df39a36e81570257ab9425131d02b9d\nDepends-On: https://review.opendev.org/c/openstack/networking-bgpvpn/+/883057\nRelated-Bug: 2019097\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/23a3f6026e5c0a2026804e71b17c458751c7844e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/23a3f6026e5c0a2026804e71b17c458751c7844e"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
