)]}'
{"id":"openstack%2Fnetworking-bagpipe~891325","triplet_id":"openstack%2Fnetworking-bagpipe~master~I5633d067e6ae4d9de89888a2587470715cd7b506","project":"openstack/networking-bagpipe","branch":"master","topic":"bug/2025127","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-09-08 08:44:49.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-09-08 11:06:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5633d067e6ae4d9de89888a2587470715cd7b506","subject":"SQLAlchemy: add context wrappers to SFC driver","status":"MERGED","created":"2023-08-14 14:16:33.000000000","updated":"2023-09-08 11:07:06.000000000","submitted":"2023-09-08 11:06:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"891325-bug/2025127","meta_rev_id":"4af7faa88721ee5314a49de86881b8e34d71a49d","_number":891325,"virtual_id_number":891325,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-08 11:06:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-09-08 08:44:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-09-08 09:41:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-09-08 08:44:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-09-08 08:44:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-14 14:52:37.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-08-14 16:01:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-08 09:41:53.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"3c77332fcca6f1f0c08a516e27a0beba759a43a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-14 14:16:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92b9ec31aa649599617701c93a4bf6eaecb77ed7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-14 14:52:37.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5d271d09f065242661929d5af614b8b7cf421809","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 16:01:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b021def88db744e986500059d6169cdb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1840bd76f93148ae8288e4c4a3963cc5 : FAILURE in 7m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/977c21f9e1b3444f9af32137e14fd40b : FAILURE in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b00862bb3e9946228c05dc8deed51efa : FAILURE in 7m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ee94d662775c4c8786af495826adf80d : FAILURE in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ab99399aa334ae597b6012b5f84e584 : SUCCESS in 11m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2099cb713a7d4c3fa6f3b4ad0361cd28 : FAILURE in 8m 18s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/ebcfc80408734ec19545a71cf234a25e : SUCCESS in 1h 37m 34s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/546e82c6c1574d48b8a800e29a289e1e : FAILURE in 10m 27s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/85b686e19b3f411b84f5c24f2b7e0ae6 : SUCCESS in 1h 30m 14s","accounts_in_message":[],"_revision_number":1},{"id":"9b5076307313a405978319791194f0353c1b206f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-15 07:39:49.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e1e3dcbb72954d7c5878f91f180a27a269e36842","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-15 07:40:17.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8d206d18a1e38b8621b1fa99e1e53cfcd154cfec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 09:23:33.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/74679910910f4fd5b665c7cab8bee71e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7aa799fa423d4bfe9bae262e638d116b : FAILURE in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6b14bd06b304cb2b9abc28143ed15eb : SUCCESS in 6m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36fe5921665d45be8559f3025a8be7bf : SUCCESS in 6m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/55fd781d790047749d7d9751f40ff19b : SUCCESS in 7m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d68c432b4bab4918b0c0b75590142c7d : SUCCESS in 11m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06da787c83414311b1e5610c3ea38897 : SUCCESS in 7m 25s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/6eeab7bbff454b5e83a39ff35032ad01 : SUCCESS in 1h 36m 29s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/09fc08826d7444d599840673ebb9bdb8 : FAILURE in 20m 18s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/7f994b22153943ddac55847f7a2dde2f : SUCCESS in 1h 05m 38s","accounts_in_message":[],"_revision_number":2},{"id":"e3e4ad19eb240943ab47b12b3ed2d3839c1f0388","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-15 13:37:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"24b79215038eabe3616502a159242f35fae281fe","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 14:01:47.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f4586faa85a4c6c85b7bad4a2598743\n\n- openstack-tox-py310-with-oslo-master https://zuul.opendev.org/t/openstack/build/ad38d2b8f7c647eaa593956f9a8ef554 : SUCCESS in 14m 07s\n- networking-bagpipe-openstack-tox-py310-with-sqlalchemy-main https://zuul.opendev.org/t/openstack/build/b13f59de9e694fd0943ca18234579c48 : FAILURE in 15m 25s","accounts_in_message":[],"_revision_number":2},{"id":"eca704a97be3548b044d2890c2c3c1a67578db5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-18 13:42:32.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b12e3771b2ec49ce05d0f0cec515ecc4c1c9ed89","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-18 13:43:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1661e8015c8ce4714eed9df9049e40000918d4a1","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 14:03:32.000000000","message":"Patch Set 3:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fafef4277564d389e5ab3073f5d2f7c\n\n- openstack-tox-py310-with-oslo-master https://zuul.opendev.org/t/openstack/build/856a757aef0647ec9a1eaa1ac5edc0cd : SUCCESS in 15m 10s\n- networking-bagpipe-openstack-tox-py310-with-sqlalchemy-main https://zuul.opendev.org/t/openstack/build/ab7c1392fa0042acbd7bc8805c89961a : FAILURE in 18m 05s","accounts_in_message":[],"_revision_number":3},{"id":"54f4448000e12885e99989de25619f034780e4fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 15:27:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd6fc93efac6415680c45df29d550048\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d71efe2250f240148e6f5c73133125f3 : SUCCESS in 7m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11c6e66c21ee4d0f9c0279bce017f7d1 : SUCCESS in 8m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce1a1ec6d5d84173af12516e73b31386 : SUCCESS in 8m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1e386b1b37324520ae3942e3603e069f : SUCCESS in 8m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/417ead064adb4f3dbcaa856e735df78d : RETRY_LIMIT in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c8b9761585a44b4a32dae6aeecb051f : SUCCESS in 11m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6aa67c5054d14ee9ad6617bf86188c91 : SUCCESS in 8m 46s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/770164ba90f2441c8f873bc4a23c715b : SUCCESS in 1h 38m 50s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/d3ebaea2397c4dd8ab79be3fc847b091 : FAILURE in 20m 47s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/1bb6e3e230f949458a2ceca2d32f1597 : SUCCESS in 1h 15m 10s","accounts_in_message":[],"_revision_number":3},{"id":"f1bc14a79fef4c6a83a148df76ed2f742198cb97","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-22 11:14:18.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c3c9552ebac614f0855388a613a8cb466713d532","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 11:35:25.000000000","message":"Patch Set 3:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e0c330887b444dbaca65b7f5eb431ec\n\n- openstack-tox-py310-with-oslo-master https://zuul.opendev.org/t/openstack/build/4e53bf92321d4f0dae6a31e13202bdd3 : SUCCESS in 13m 04s\n- networking-bagpipe-openstack-tox-py310-with-sqlalchemy-main https://zuul.opendev.org/t/openstack/build/e3763cc603fc402689ef5a995d82194b : FAILURE in 15m 44s","accounts_in_message":[],"_revision_number":3},{"id":"2b27c051688187bafa659df77556402ab97e6caa","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-22 12:40:43.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"25fd49fd313fd1866a5daddb8bd87b91171ba548","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-22 15:17:45.000000000","message":"Removed Code-Review+2 by \u003cGERRIT_ACCOUNT_16688\u003e\n","accounts_in_message":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"_revision_number":3},{"id":"22d42efc9a457cfc80d16c6e520c8f7975f9f2fd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-22 15:25:59.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"068a5ae0e0046211acc5c3eda6600bf80921e2f2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-22 16:03:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b9e8078c7123abc26e52dc6928773f345bdd8bc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-23 12:20:42.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"276751a70a2629ca79e7664e8be84a4491097be7","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-23 12:22:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f29c2d3580952d653aa2de7ba48a239fbb5ee839","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 12:39:07.000000000","message":"Patch Set 4:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4e1a590d1d84aa1bf845f621e0565d0\n\n- openstack-tox-py310-with-oslo-master https://zuul.opendev.org/t/openstack/build/201c6a4c0a544d938768ff2c08304184 : SUCCESS in 11m 49s\n- networking-bagpipe-openstack-tox-py310-with-sqlalchemy-main https://zuul.opendev.org/t/openstack/build/87208d5d6c2748b9a87caf1c56c9394d : FAILURE in 14m 41s","accounts_in_message":[],"_revision_number":4},{"id":"aac4393d883b1c66dc06710478d19b3d45723f58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 14:04:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88ac90c9ca874e02a7851499554ca0ca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99b687c727a240b28b2abe1443a25824 : SUCCESS in 9m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/729553e5d269443b81d6a1fcd5afba73 : SUCCESS in 7m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b1340b1850d48b8a4348122566f6475 : SUCCESS in 7m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e75a87595d3241de9a2af99478d56ca8 : SUCCESS in 8m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f8b8937f86184ded88ad3e4803bab1bb : RETRY_LIMIT in 2m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2add965a593f448ebc0f8d426c883845 : SUCCESS in 10m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb8f216c59e245099f41c457131fc5f8 : SUCCESS in 8m 23s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/c7606cd4b2ca465ca1e56d02eb61d7f0 : SUCCESS in 1h 17m 48s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/ff2f3d5b08a447ed888a035a99655d42 : FAILURE in 19m 59s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/40d133c1824e40b79c7a4beabd846e79 : SUCCESS in 1h 43m 24s","accounts_in_message":[],"_revision_number":4},{"id":"5024dbf5b818d5da8dec0a50ec7d2815eba4bfe9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-05 08:43:59.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":4},{"id":"1fb44923d867b633e6eee7988ee51a296a15de7f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-05 08:44:29.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"89980422ec8f2adca3bfeac41e8ff460e02857d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 08:44:59.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"70d9926678dbb3208f1626eaea71a4f91c01a113","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 09:47:32.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/da1d13aa8b834c6db550abb7281a6ef5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1ac6905c40540fda88475058242d7f8 : SUCCESS in 9m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51680c1ecd874e5d8b06377aee255222 : SUCCESS in 8m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ca715380d3448518054ff4252c7e87d : SUCCESS in 8m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/514b43ea82e14b979ccd2f8cfbd0a62e : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d127a071b844969b6697181264bc41d : SUCCESS in 9m 37s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/c33f6d4d718547bfb36d97986b85aed8 : SUCCESS in 1h 01m 46s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/885dd4efb9114e37a5d3fcf386c134d3 : FAILURE in 5m 26s","accounts_in_message":[],"_revision_number":4},{"id":"df408482bdf3af05abae25f415c880eba120c4eb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-05 15:04:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ae22444535d24d743bd9aa490d52b870d085ada7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 16:22:07.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7458f4b71d124a56a4d75659545de9cb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5d817bf3ef84dc3ab0b5c0c4f1df01d : SUCCESS in 7m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a452efdd0a242e38cd509213e992a17 : SUCCESS in 8m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d91fa7214e54c049f3d9c90e70c60b7 : SUCCESS in 7m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f76378b2ba264b9eadd59331be5dfd98 : SUCCESS in 7m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5b1e5dcc6cbe4752924f26056cce1b6b : RETRY_LIMIT in 2m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f96abcf7d1349358bf5fc3fc1495b64 : SUCCESS in 11m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e815ef4ebeb47f092025f63da90c5ba : SUCCESS in 8m 03s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/f1b3f433f22b48aa8b77e5bbcc2df39e : SUCCESS in 1h 16m 43s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/d5efaf3713c543f6b583ec1cfdea0cc4 : FAILURE in 8m 16s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/2addb1216cce47ad83910c05e1f1748c : FAILURE in 15m 14s","accounts_in_message":[],"_revision_number":4},{"id":"2d99fcc0459835725fbed15506341714320174a0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-06 08:45:55.000000000","message":"Topic bug/1964575 removed","accounts_in_message":[],"_revision_number":4},{"id":"75e926131762bc3382f34b2022f5471e092390a9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-06 08:46:03.000000000","message":"Topic set to bug/2025127","accounts_in_message":[],"_revision_number":4},{"id":"2620c19bd72f17fe58404dce813c37b02477f40c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-07 18:37:56.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":5},{"id":"0dc90027e8c5d852508fcdf07bde07bfe7672e06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 20:23:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/420904fa60914555951f89c73edbb9b4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bebb38d79584e388bfe0d55b9e9e5fe : SUCCESS in 8m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/514195892c694bd09cb2f748e433e6da : SUCCESS in 8m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad7740d9415241c3b67bd7c60c72c19e : SUCCESS in 7m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/296e3b4d18e24d1b80bf797bde7db292 : SUCCESS in 7m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/430bb8dfd0a34bff9062343d7bc9912d : RETRY_LIMIT in 1m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee6e7bcca421423fbdf651a9dab3cf07 : SUCCESS in 8m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f9f03b255644d7ca56c5ac6c7010424 : SUCCESS in 8m 12s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/434351641bc44eb4be905d181261a383 : SUCCESS in 1h 39m 11s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/13ea7120f34548b9ab837dd9fa49c87f : FAILURE in 8m 53s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/219352029dbf4bb292f760ad3307f868 : SUCCESS in 1h 37m 27s","accounts_in_message":[],"_revision_number":5},{"id":"d46adfc63226bb89df3083411d98312c38bbea74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-08 08:06:57.000000000","message":"Patch Set 6: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"e6788eee2cb17473fb93272e121f50c904fc662b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-08 08:44:49.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1 Review-Priority+2","accounts_in_message":[],"_revision_number":6},{"id":"665eff652b8f239009327719c363e5ffbda94f3e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-09-08 09:41:53.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"9c9b8998109698aababd94c2f5f09716c02fd59f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 09:42:36.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3c5fbf0ac5145aa91b4d54ecd40d0fd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69902b36600d421985ceea16676781f7 : SUCCESS in 8m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69d1b599450c44799cbc908224b3915a : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7263a2be6f6a4c99b43b1bd9b8430042 : SUCCESS in 6m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cecb4dc77c9646e09fdba4ac8228e357 : SUCCESS in 6m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7e5611c50963408897cb9d44c5903e54 : RETRY_LIMIT in 2m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be61afcf95e6426395cd8f168d826adc : SUCCESS in 12m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45e52d9fc254456eb25a51326595fa87 : SUCCESS in 9m 09s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/06ca7a7d070f4c209e219893758ec83b : SUCCESS in 1h 30m 12s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/cdf2db4266324d8b97436f97fa99b73c : FAILURE in 10m 47s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/046d8f3a7aaf4e11ab80c0184754dc95 : SUCCESS in 1h 20m 14s","accounts_in_message":[],"_revision_number":6},{"id":"fe5fe9302f10c49f8575fe4edaa1010652427cbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 09:43:08.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"83cc4093c52633a956b4c67394c696c83a85d51b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 11:06:09.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4edc3c9b0bf24630887d36701013d8bc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06bbaa8791cd4165bf39215646230166 : SUCCESS in 8m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/adc66712d4b84045a8bde8bd82a41046 : SUCCESS in 7m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57ddd2fc8845443fb961ec24f4af35e4 : SUCCESS in 8m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b20b2c50b60545ed89c4842367c58550 : SUCCESS in 8m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5b055d8b0fb428c8902fecdd06b12c6 : SUCCESS in 12m 21s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/7f3df27cbe4f4ec8a8d9a634a4690ab3 : SUCCESS in 1h 21m 50s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/e6df74a3219b4082b8290de56593df7f : SUCCESS in 1h 14m 49s","accounts_in_message":[],"_revision_number":6},{"id":"8b3c38d250b211bb7b27295fe7916d9e6c51630f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 11:06:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"4af7faa88721ee5314a49de86881b8e34d71a49d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 11:07:06.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a51fc2a79274a3c94198460049789f2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13662a40f36c4a5fb0e9a5540f30fba7 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"58efb3e21cce6b1661782377b08dfe5a2257788b","revisions":{"1046087fe5db1dba7b79a4d0d2495dc9f096d4b1":{"kind":"REWORK","_number":1,"created":"2023-08-14 14:16:33.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/25/891325/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/25/891325/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/25/891325/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/25/891325/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/25/891325/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/25/891325/1"}}},"commit":{"parents":[{"commit":"c8a974c435ea48219862cf9e6a34f9a95fe24675","subject":"[sqlalchemy-20]: remove subtransactions\u003dTrue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c8a974c435ea48219862cf9e6a34f9a95fe24675"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-08-14 14:07:19.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-08-14 14:07:19.000000000","tz":120},"subject":"SQLAlchemy: add context wrappers to SFC driver","message":"SQLAlchemy: add context wrappers to SFC driver\n\nWith SQLAlchemy \u003e 2.0.0 the unit tests of networking-bagpipe\nare unstable.\nThe failure looks like some session violance, but more work\nis necessary as I still can\u0027t see where the root of the issue is.\n\nChange-Id: I5633d067e6ae4d9de89888a2587470715cd7b506\nRelated-Bug: #1964575\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/1046087fe5db1dba7b79a4d0d2495dc9f096d4b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/1046087fe5db1dba7b79a4d0d2495dc9f096d4b1"}]},"branch":"refs/heads/master"},"11cf0978b653ad18b6b16f79856e78676a4a62b6":{"kind":"REWORK","_number":2,"created":"2023-08-15 07:39:49.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/25/891325/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/25/891325/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/25/891325/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/25/891325/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/25/891325/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/25/891325/2"}}},"commit":{"parents":[{"commit":"c8a974c435ea48219862cf9e6a34f9a95fe24675","subject":"[sqlalchemy-20]: remove subtransactions\u003dTrue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c8a974c435ea48219862cf9e6a34f9a95fe24675"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-08-14 14:07:19.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-08-15 07:39:30.000000000","tz":120},"subject":"SQLAlchemy: add context wrappers to SFC driver","message":"SQLAlchemy: add context wrappers to SFC driver\n\nWith SQLAlchemy \u003e 2.0.0 the unit tests of networking-bagpipe\nare unstable.\nThe failure looks like some session violance, but more work\nis necessary as I still can\u0027t see where the root of the issue is.\n\nChange-Id: I5633d067e6ae4d9de89888a2587470715cd7b506\nRelated-Bug: #1964575\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/11cf0978b653ad18b6b16f79856e78676a4a62b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/11cf0978b653ad18b6b16f79856e78676a4a62b6"}]},"branch":"refs/heads/master"},"0c3e54c667f9dd54a1de9c28667c7d967eed1354":{"kind":"REWORK","_number":3,"created":"2023-08-18 13:42:32.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/25/891325/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/25/891325/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/25/891325/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/25/891325/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/25/891325/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/25/891325/3"}}},"commit":{"parents":[{"commit":"c8a974c435ea48219862cf9e6a34f9a95fe24675","subject":"[sqlalchemy-20]: remove subtransactions\u003dTrue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c8a974c435ea48219862cf9e6a34f9a95fe24675"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-08-14 14:07:19.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-08-18 13:42:14.000000000","tz":120},"subject":"SQLAlchemy: add context wrappers to SFC driver","message":"SQLAlchemy: add context wrappers to SFC driver\n\nWith SQLAlchemy \u003e 2.0.0 the unit tests of networking-bagpipe\nare unstable.\nThe failure looks like some session violance, but more work\nis necessary as I still can\u0027t see where the root of the issue is.\n\nChange-Id: I5633d067e6ae4d9de89888a2587470715cd7b506\nRelated-Bug: #1964575\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0c3e54c667f9dd54a1de9c28667c7d967eed1354"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0c3e54c667f9dd54a1de9c28667c7d967eed1354"}]},"branch":"refs/heads/master"},"aad96eb8e3dbb8069ff6325bc62e139247a219c1":{"kind":"REWORK","_number":4,"created":"2023-08-23 12:20:42.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/25/891325/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/25/891325/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/25/891325/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/25/891325/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/25/891325/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/25/891325/4"}}},"commit":{"parents":[{"commit":"c8a974c435ea48219862cf9e6a34f9a95fe24675","subject":"[sqlalchemy-20]: remove subtransactions\u003dTrue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c8a974c435ea48219862cf9e6a34f9a95fe24675"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-08-23 12:20:02.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-08-23 12:20:02.000000000","tz":120},"subject":"SQLAlchemy: add context wrappers to SFC driver","message":"SQLAlchemy: add context wrappers to SFC driver\n\nWith SQLAlchemy \u003e 2.0.0 the unit tests of networking-bagpipe\nare unstable.\nThe failure looks like some session violance, but more work\nis necessary as I still can\u0027t see where the root of the issue is.\n\nChange-Id: I5633d067e6ae4d9de89888a2587470715cd7b506\nRelated-Bug: #1964575\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/aad96eb8e3dbb8069ff6325bc62e139247a219c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/aad96eb8e3dbb8069ff6325bc62e139247a219c1"}]},"branch":"refs/heads/master"},"e501c5d92bb5e3ac431d555aef29d625ecc287c9":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-09-07 18:37:56.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/25/891325/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/25/891325/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/25/891325/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/25/891325/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/25/891325/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/25/891325/5"}}},"commit":{"parents":[{"commit":"60d374d8a75e5c2c5ef3053c07b09ac43ac0790e","subject":"CI: Change focal nodeset to jammy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/60d374d8a75e5c2c5ef3053c07b09ac43ac0790e"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-08-23 12:20:02.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2023-09-07 18:37:56.000000000","tz":0},"subject":"SQLAlchemy: add context wrappers to SFC driver","message":"SQLAlchemy: add context wrappers to SFC driver\n\nWith SQLAlchemy \u003e 2.0.0 the unit tests of networking-bagpipe\nare unstable.\nThe failure looks like some session violance, but more work\nis necessary as I still can\u0027t see where the root of the issue is.\n\nChange-Id: I5633d067e6ae4d9de89888a2587470715cd7b506\nRelated-Bug: #1964575\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e501c5d92bb5e3ac431d555aef29d625ecc287c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e501c5d92bb5e3ac431d555aef29d625ecc287c9"}]},"branch":"refs/heads/master","description":"Rebase"},"58efb3e21cce6b1661782377b08dfe5a2257788b":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2023-09-08 08:06:57.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/25/891325/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/25/891325/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/25/891325/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/25/891325/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/25/891325/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/25/891325/6"}}},"commit":{"parents":[{"commit":"60d374d8a75e5c2c5ef3053c07b09ac43ac0790e","subject":"CI: Change focal nodeset to jammy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/60d374d8a75e5c2c5ef3053c07b09ac43ac0790e"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-08-23 12:20:02.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2023-09-08 08:06:57.000000000","tz":0},"subject":"SQLAlchemy: add context wrappers to SFC driver","message":"SQLAlchemy: add context wrappers to SFC driver\n\nWith SQLAlchemy \u003e 2.0.0 the unit tests of networking-bagpipe\nare unstable.\nThe failure looks like some session violance, but more work\nis necessary as I still can\u0027t see where the root of the issue is.\n\nChange-Id: I5633d067e6ae4d9de89888a2587470715cd7b506\nRelated-Bug: #1964575\nRelated-Bug: #2025127\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/58efb3e21cce6b1661782377b08dfe5a2257788b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/58efb3e21cce6b1661782377b08dfe5a2257788b"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
