)]}'
{"id":"openstack%2Fnetworking-bagpipe~894260","triplet_id":"openstack%2Fnetworking-bagpipe~master~I31fdd65eabf84c0fba1ae4db834d67c95c7c26a2","project":"openstack/networking-bagpipe","branch":"master","topic":"bug/2004265","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-09-08 11:36:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I31fdd65eabf84c0fba1ae4db834d67c95c7c26a2","subject":"Unit tests: replace the deprecated assertDictContainsSubset call","status":"MERGED","created":"2023-09-08 08:05:11.000000000","updated":"2023-09-08 11:36:55.000000000","submitted":"2023-09-08 11:36:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894260-bug/2004265","meta_rev_id":"3747b8e816a5b19aba9605e5812821d21f70e83a","_number":894260,"virtual_id_number":894260,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-08 11:36:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-09-08 08:45:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-09-08 09:41:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-09-08 09:41:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-08 08:45:46.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-09-08 09:41:59.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-09-08 09:58:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5ea5250e9bde73f41b9d943a929e2654bbab5dc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-08 08:05:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ce4e3fb67eaeb702dd1e7f41f9354342f49f5ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-08 08:07:42.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"27e3ecb182899bea10fc8d63f8c99caaa7950da4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-08 08:45:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7b846f76cee74ba7a2a59235ec93a0410436ac86","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-09-08 09:41:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c6567fcb0c9b5fbe17c50beaea2e8278cef7ffcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 09:58:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fea7cdf378d4098b46bc4c0f108a171\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b5510f1b50b46e082bfb5165bb50daa : SUCCESS in 8m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed5b1660510d4034929885afdee95bed : SUCCESS in 7m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b53df87fa7248d88028e3166e8a63f4 : SUCCESS in 8m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/695c46ed207640e98220a3435ecccfbe : SUCCESS in 7m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e331b0992cd040c890b808a61efcf221 : RETRY_LIMIT in 1m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d0bf70311db4d68950625e1bb005775 : SUCCESS in 11m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a71ad8f087c4620b6273220e0823a78 : SUCCESS in 8m 28s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/02a2f14cc292431c9406b9737ac07234 : SUCCESS in 1h 41m 39s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/2720d2fa23d84bb88c4639b5f9881b81 : FAILURE in 19m 21s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/c9014cf6a94743c3bbc35348715f74ad : SUCCESS in 1h 22m 43s","accounts_in_message":[],"_revision_number":2},{"id":"78a4237ad0b2129d838651360ff54e4da0912b65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 09:59:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"70481619831c6a16f600245121ee604092a9da1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 11:36:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f80fc9b11b74c0a8b93a2df58a66369\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c5e67b64c674693acbb95e88ca56cdd : SUCCESS in 8m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d80be3b6cc13490ea1cbde4c6ab145a2 : SUCCESS in 9m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/735d69207edd4a1a99a294f29b60c5c1 : SUCCESS in 7m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c2e5ef3f2f64e7599687999b749ef3f : SUCCESS in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e42056dfed7a4fc88e39367c983358bd : SUCCESS in 10m 10s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/bc72f0387c064ef0a3a7d42e21454801 : SUCCESS in 1h 26m 57s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/cc00203cd6cd4fb5a26761095f10e533 : SUCCESS in 1h 29m 58s","accounts_in_message":[],"_revision_number":2},{"id":"60d713b0ff83961f124c29fbd84f19accfc0e1f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 11:36:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3747b8e816a5b19aba9605e5812821d21f70e83a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 11:36:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f350aaa1c138415d930da3ab69e475b0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03c973c4099e4666816e2ff8bf4b967d : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"09efb169ff926e351135f01b88fdfb3a08581811","revisions":{"d8f40f55228a30a88ff1c97a821c7ecfe5de8ad7":{"kind":"REWORK","_number":1,"created":"2023-09-08 08:05:11.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/60/894260/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/60/894260/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/60/894260/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/60/894260/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/60/894260/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/60/894260/1"}}},"commit":{"parents":[{"commit":"cbc43b6af0576d04c5bfecb34a4adac2f7317a08","subject":"[sqlalchemy-20] Add the remaining context wrappers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/cbc43b6af0576d04c5bfecb34a4adac2f7317a08"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-09-08 07:49:11.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-09-08 07:50:25.000000000","tz":120},"subject":"Unit tests: replace the deprecated assertDictContainsSubset call","message":"Unit tests: replace the deprecated assertDictContainsSubset call\n\nassertDictContainsSubset was deprecated in python 3.2 see [1].\n\n[1]: https://docs.python.org/3.2/library/unittest.html#unittest.TestCase.assertDictContainsSubset\n\nChange-Id: I31fdd65eabf84c0fba1ae4db834d67c95c7c26a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/d8f40f55228a30a88ff1c97a821c7ecfe5de8ad7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/d8f40f55228a30a88ff1c97a821c7ecfe5de8ad7"}]},"branch":"refs/heads/master"},"09efb169ff926e351135f01b88fdfb3a08581811":{"kind":"NO_CHANGE","_number":2,"created":"2023-09-08 08:07:42.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/60/894260/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/60/894260/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/60/894260/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/60/894260/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/60/894260/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/60/894260/2"}}},"commit":{"parents":[{"commit":"48f12c0019c5088a10b7c74c78eb821d29b7f402","subject":"[sqlalchemy-20] Add the remaining context wrappers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/48f12c0019c5088a10b7c74c78eb821d29b7f402"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-09-08 07:49:11.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2023-09-08 08:07:42.000000000","tz":0},"subject":"Unit tests: replace the deprecated assertDictContainsSubset call","message":"Unit tests: replace the deprecated assertDictContainsSubset call\n\nassertDictContainsSubset was deprecated in python 3.2 see [1].\n\n[1]: https://docs.python.org/3.2/library/unittest.html#unittest.TestCase.assertDictContainsSubset\n\nChange-Id: I31fdd65eabf84c0fba1ae4db834d67c95c7c26a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/09efb169ff926e351135f01b88fdfb3a08581811"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/09efb169ff926e351135f01b88fdfb3a08581811"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
