)]}'
{"id":"openstack%2Fnetworking-bagpipe~900216","triplet_id":"openstack%2Fnetworking-bagpipe~master~I7af2b438960233a9140aa711c27c330c58ebb176","project":"openstack/networking-bagpipe","branch":"master","topic":"pip-23.1-support","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-09-03 21:49:14.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-09-03 21:49:14.000000000","reason":"Change was submitted"},"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2024-08-30 18:03:17.000000000","reason":"\u003cGERRIT_ACCOUNT_9656\u003e replied on the change","reason_account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-09-03 21:49:14.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-09-03 20:07:33.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I7af2b438960233a9140aa711c27c330c58ebb176","subject":"Add pyproject.toml to support pip 23.1","status":"MERGED","created":"2023-11-06 16:56:46.000000000","updated":"2024-09-03 21:50:13.000000000","submitted":"2024-09-03 21:49:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"900216-pip-23.1-support","meta_rev_id":"d64691eac037028eacee5bfb025d1d8326b613da","_number":900216,"virtual_id_number":900216,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-03 21:49:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"all":[{"value":2,"date":"2024-08-30 18:03:17.000000000","_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":2,"date":"2024-09-03 20:07:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":1,"date":"2024-09-03 20:07:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-06 18:31:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-17 20:25:51.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"CC"},{"updated":"2024-08-30 18:03:17.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2024-09-02 09:06:27.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-09-02 09:06:27.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-09-02 09:06:27.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"2a5a1d501144afa36bcdce486700c0d7ac322600","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-11-06 16:56:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b33e8a658be053197f0c6d6bd5b418ca74688fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-11-06 17:01:57.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"82c8145a1102977de68170a8a40d1000e29e7612","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 18:31:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/db8d8e1dc1a345728c23207e86b59b12\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4df90e5f0d684cf4ba34c1e16492c59d : SUCCESS in 9m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8ba0dfd581f4d488ee602f760581b38 : SUCCESS in 9m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/205039198fb646c4bca973d02d6c9298 : FAILURE in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f6d2124f42f9404b8f77168ea4e05815 : SUCCESS in 8m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/140d83ad364244aab86a91a47ea3b4f1 : SUCCESS in 9m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f12d4e13947a472ba9569f5ccb3a8b27 : SUCCESS in 12m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0aa26ea2d1a4aa6bef62c615c2533b6 : SUCCESS in 13m 40s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/cfe4fe9566a145429af8b9ef68395f32 : SUCCESS in 1h 20m 22s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/ce9cdd0e7b50427e8714e3d63b8e83b3 : FAILURE in 20m 45s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/ddb6ccf8ae104538b94f6b2f203618ad : SUCCESS in 1h 29m 12s","accounts_in_message":[],"_revision_number":2},{"id":"4028dedacb950811badef17adedb1b56e80f6e90","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-11-07 13:19:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8bb02f559f1093acc465931c1ef83be222bd1f34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 15:13:56.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/68baea8bbe574dc4a6e32662a5224f96\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c757e673ea2b405b80da481bcd254604 : SUCCESS in 9m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/439629ca73474f60bd2204d739aeffe1 : SUCCESS in 8m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a06878cb430486380c8ae1536bb4816 : FAILURE in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1979edc877b84322baf436a668a4e6f6 : SUCCESS in 7m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fb948cedea624c1e85b0f4163213cb31 : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/350127b58f5e473aa4cd6313999ef23f : SUCCESS in 13m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d340d39f4ad4cf293c9f41b4256e047 : SUCCESS in 8m 39s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/52b99a3d77554195b6d48e1eee4d2467 : SUCCESS in 1h 40m 22s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/066cfcd2e60542c487e4d5d47ca129a5 : FAILURE in 21m 09s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/f257f963fcc447b1934bf6ec416a466d : SUCCESS in 1h 45m 41s","accounts_in_message":[],"_revision_number":2},{"id":"b5eb1492e4e426e39f03538c3b301e134dcf1fce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-11-09 15:45:52.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"326a609718391d1b59584b48cdbdf383e4a20cf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 17:27:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42e931ed48134796a81c6251d8554c8b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7195b9de82142e28df99af804010c56 : SUCCESS in 7m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39ede12a09f14239a8ad0eb61fbf7daa : SUCCESS in 7m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84472fe610f24d5286fffde6418002e2 : SUCCESS in 7m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b9b3645f56df428f8417987db51bb814 : SUCCESS in 8m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f158774a9b64466868b16431f3ba914 : SUCCESS in 10m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb2d6fba665744f88e422f9416b039d6 : SUCCESS in 9m 17s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/62048637f3094f56ba47b9855eaea27a : SUCCESS in 1h 28m 19s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/8913857634ed40f892d634e9ebbf6a36 : FAILURE in 21m 52s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/bcd68ec5f2ef4029bc29fa8e2b6d19b7 : SUCCESS in 1h 40m 38s","accounts_in_message":[],"_revision_number":3},{"id":"61a5ac8ab1cddd811d9a924d73bc80d6238e16af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-04-19 11:13:19.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"d69a90c8535adbfd2c5aecc532f6b44776a1da82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 13:24:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c0d20c109f04bac9ff8ae0527711289\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3db799f0e5ff416a882923566dcdb1f8 : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7da28752b37945778e091985403dc407 : SUCCESS in 7m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b22681771007469a9c9412e23a94e799 : SUCCESS in 6m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca67f202e3774dcb93e33221848c6c0e : SUCCESS in 6m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7b5620a999274defb9a617f65023ad6d : SUCCESS in 8m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9cd838d3035f425c81bda4891164593d : FAILURE in 11m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/776c9638daca40bba52b76d53fd8ef6f : SUCCESS in 10m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86d198767dec447893441444f018db39 : SUCCESS in 7m 48s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/a7f1c0c2932c4177b03f2116f15ea5a6 : TIMED_OUT in 2h 02m 35s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/2031ac9893e8434196b59aa23ce9a624 : FAILURE in 15m 55s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/06c5bbae69204d5a94d2e3f595212851 : SUCCESS in 1h 14m 26s\n\nWarning:\n  Comments left for invalid file .pylintrc","accounts_in_message":[],"_revision_number":4},{"id":"8a35840d3edb3c686742f09f12f7b4516c7fabf5","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-04-22 11:46:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f9f81c5ba83f02865a245e184cb8415717deaeb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 13:22:59.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f918e32e73248ffbe912328cf29c92e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e90df6e7bb484cf4aa693cc33566a6bf : SUCCESS in 7m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1cd107baccc8442f9101b969ad0ce019 : SUCCESS in 7m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c757edc1d63f4352beac5b4ab189c556 : SUCCESS in 6m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e878bd6c37a4ae5a3ecf25ff1ca13b2 : SUCCESS in 8m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6d013f7eecc44783bdd4388063268321 : SUCCESS in 8m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ff332fdb734b468583e63c4fc7e08152 : FAILURE in 9m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4278c4340fe2429caa4ec35b0ba0aabb : SUCCESS in 12m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c69d47b3eb6344e59411e23636c55d73 : SUCCESS in 9m 28s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/0f1f9b2e00df483eb7cb0ca465b9699d : FAILURE in 1h 22m 59s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/44d79312b7804dd3bfcc59be9aa9b558 : FAILURE in 8m 27s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/814a332666b44c839fc61e66bf82a44c : SUCCESS in 1h 26m 01s\n\nWarning:\n  Comments left for invalid file .pylintrc","accounts_in_message":[],"_revision_number":4},{"id":"28c93e74db7a1a92fec1b5f124668cb279054cb1","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-17 20:25:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4ece3ec3ae1903009095befdfd750aba0f0772b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-07-04 08:13:53.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"37da3f892db538dacac78651b2172fe9aaae8c51","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-07-04 08:14:34.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9cab9d22be7e6af10236672c930872aa5d2018f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 10:14:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73eccabbd2f049f2afef94d821c76bad\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6edff521ecdf409a8cc8feb50337ff11 : SUCCESS in 8m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b95577006bc84c64877ec4dc9bbc7d54 : SUCCESS in 7m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/32faaadd7f6449b7940883cf97417914 : SUCCESS in 6m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ba4bcdcd46574bbf890a3944aca44e1a : SUCCESS in 6m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/69f03f7f3dd345999652b43e96be3183 : SUCCESS in 12m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d76a0c64eebd445890f761413b45ba66 : SUCCESS in 11m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/236686cfb8f64020b636aaf388ec8874 : SUCCESS in 8m 04s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/e748698dc6a54a35a5a1a27a4c49af1a : SUCCESS in 1h 54m 14s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/57c9fc546abb4db2b33469fa038ffc10 : FAILURE in 10m 16s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/7d4795df06c0480595319d7e71cacd14 : SUCCESS in 1h 41m 00s\n\nWarning:\n  Comments left for invalid file .pylintrc","accounts_in_message":[],"_revision_number":5},{"id":"7a28a6264353a0636c23d96ffee8e2b5c4f317ce","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-08-30 18:03:17.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"dc0232a675dbe2c5cb6ce0db87c83c217a94d77b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-09-03 20:07:33.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"6aeaacaaaf56b9eec858e4f291fc3284f6693828","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 20:08:04.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"fa8fd6d9f7325e98eb8552a7721d24060d007972","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 21:49:14.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4e4eb38ec7d47dabb3c8ce3c4046c18\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/148e33b5c4fa4ffab59f467775336329 : SUCCESS in 7m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8957b682120045499edb1772b75c2533 : SUCCESS in 9m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4c2c03801a214278b92dce0fd364377a : SUCCESS in 8m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b99989d21ddd4d908c235b7988d5340f : SUCCESS in 9m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7382af38abdd4614adf6f995635531cf : SUCCESS in 11m 16s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/341ae68feac1474a9ed20936595f610a : SUCCESS in 1h 39m 41s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/c3e9ab4dc2394c19a458f6006317b2a1 : SUCCESS in 1h 23m 18s\n\nWarning:\n  Comments left for invalid file .pylintrc","accounts_in_message":[],"_revision_number":5},{"id":"7be94f46d49892cf09586fb76e232877a9d8d0ba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 21:49:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"d64691eac037028eacee5bfb025d1d8326b613da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 21:50:13.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3b8397df71e459fae13edf240aa2a96\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20fd319ad5c74923be6e8d9a9ad2e274 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"7857f51885e87e334911b479c6e0ec8d9c97726b","revisions":{"f853656f45455d07dc1cf1f4fd8b09ab6d6af4cb":{"kind":"REWORK","_number":1,"created":"2023-11-06 16:56:46.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/16/900216/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/16/900216/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/16/900216/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/16/900216/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/16/900216/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/16/900216/1"}}},"commit":{"parents":[{"commit":"e8bfaf4f75fbc3b8b3376028a6d483b3fb2d2711","subject":"Merge \"Update master for stable/2023.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e8bfaf4f75fbc3b8b3376028a6d483b3fb2d2711"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-11-06 16:38:37.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-11-06 16:38:37.000000000","tz":60},"subject":"Add pyproject.toml to support pip 23.1","message":"Add pyproject.toml to support pip 23.1\n\nChange-Id: I7af2b438960233a9140aa711c27c330c58ebb176\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f853656f45455d07dc1cf1f4fd8b09ab6d6af4cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f853656f45455d07dc1cf1f4fd8b09ab6d6af4cb"}]},"branch":"refs/heads/master"},"4dbf3fcd92fdacce1c8c9a1a0ec7baace310e08d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-11-06 17:01:57.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/16/900216/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/16/900216/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/16/900216/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/16/900216/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/16/900216/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/16/900216/2"}}},"commit":{"parents":[{"commit":"e8bfaf4f75fbc3b8b3376028a6d483b3fb2d2711","subject":"Merge \"Update master for stable/2023.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e8bfaf4f75fbc3b8b3376028a6d483b3fb2d2711"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-11-06 16:38:37.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2023-11-06 17:01:56.000000000","tz":0},"subject":"Add pyproject.toml to support pip 23.1","message":"Add pyproject.toml to support pip 23.1\n\nFor details see the below mail thread:\nhttps://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/HVFN5RBSHRTM3B2UUKPAWKH6H6AT6CYR/\n\nand the commit message of the keystone patch:\nhttps://review.opendev.org/c/899519\n\nChange-Id: I7af2b438960233a9140aa711c27c330c58ebb176\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/4dbf3fcd92fdacce1c8c9a1a0ec7baace310e08d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/4dbf3fcd92fdacce1c8c9a1a0ec7baace310e08d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f5d2850cdd75fe79659c6074f79ed965773aaa29":{"kind":"REWORK","_number":3,"created":"2023-11-09 15:45:52.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/16/900216/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/16/900216/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/16/900216/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/16/900216/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/16/900216/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/16/900216/3"}}},"commit":{"parents":[{"commit":"e8bfaf4f75fbc3b8b3376028a6d483b3fb2d2711","subject":"Merge \"Update master for stable/2023.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e8bfaf4f75fbc3b8b3376028a6d483b3fb2d2711"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-11-06 16:38:37.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-11-09 15:45:40.000000000","tz":60},"subject":"Add pyproject.toml to support pip 23.1","message":"Add pyproject.toml to support pip 23.1\n\nFor details see the below mail thread:\nhttps://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/HVFN5RBSHRTM3B2UUKPAWKH6H6AT6CYR/\n\nand the commit message of the keystone patch:\nhttps://review.opendev.org/c/899519\n\nChange-Id: I7af2b438960233a9140aa711c27c330c58ebb176\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f5d2850cdd75fe79659c6074f79ed965773aaa29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f5d2850cdd75fe79659c6074f79ed965773aaa29"}]},"branch":"refs/heads/master"},"8c1dc2ed125ed23531c488abca8e4019ab8a38da":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-04-19 11:13:19.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/16/900216/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/16/900216/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/16/900216/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/16/900216/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/16/900216/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/16/900216/4"}}},"commit":{"parents":[{"commit":"385b6b4beba5cca55a4a424fdb1c63d12f4e585d","subject":"Merge \"coveragerc: Remove reference to non-existant path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/385b6b4beba5cca55a4a424fdb1c63d12f4e585d"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-11-06 16:38:37.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2024-04-19 11:13:19.000000000","tz":0},"subject":"Add pyproject.toml to support pip 23.1","message":"Add pyproject.toml to support pip 23.1\n\nFor details see the below mail thread:\nhttps://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/HVFN5RBSHRTM3B2UUKPAWKH6H6AT6CYR/\n\nand the commit message of the keystone patch:\nhttps://review.opendev.org/c/899519\n\nChange-Id: I7af2b438960233a9140aa711c27c330c58ebb176\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/8c1dc2ed125ed23531c488abca8e4019ab8a38da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/8c1dc2ed125ed23531c488abca8e4019ab8a38da"}]},"branch":"refs/heads/master","description":"Rebase"},"7857f51885e87e334911b479c6e0ec8d9c97726b":{"kind":"REWORK","_number":5,"created":"2024-07-04 08:13:53.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/16/900216/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/16/900216/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/16/900216/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/16/900216/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/16/900216/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/16/900216/5"}}},"commit":{"parents":[{"commit":"016f937ecba9a1b6e86cd730c79eb1bf7ed9a3f1","subject":"Merge \"reno: Update master for unmaintained/zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/016f937ecba9a1b6e86cd730c79eb1bf7ed9a3f1"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-11-06 16:38:37.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-07-04 08:13:32.000000000","tz":120},"subject":"Add pyproject.toml to support pip 23.1","message":"Add pyproject.toml to support pip 23.1\n\nFor details see the below mail thread:\nhttps://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/HVFN5RBSHRTM3B2UUKPAWKH6H6AT6CYR/\n\nand the commit message of the keystone patch:\nhttps://review.opendev.org/c/899519\n\nChange-Id: I7af2b438960233a9140aa711c27c330c58ebb176\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7857f51885e87e334911b479c6e0ec8d9c97726b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7857f51885e87e334911b479c6e0ec8d9c97726b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
