)]}'
{"id":"openstack%2Fnetworking-bagpipe~907189","triplet_id":"openstack%2Fnetworking-bagpipe~master~Ieaf7e926c64921b9e92cf04370bff8edb2b0b304","project":"openstack/networking-bagpipe","branch":"master","topic":"bump-hacking","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-03-06 08:05:01.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2024-02-09 15:02:34.000000000","reason":"\u003cGERRIT_ACCOUNT_9656\u003e replied on the change","reason_account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-03-06 10:01:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ieaf7e926c64921b9e92cf04370bff8edb2b0b304","subject":"Bump hacking","status":"MERGED","created":"2024-01-30 10:05:11.000000000","updated":"2024-03-06 10:02:43.000000000","submitted":"2024-03-06 10:01:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"907189-bump-hacking","meta_rev_id":"88413da67f13a4a20490df79c299f03acdcd4fd6","_number":907189,"virtual_id_number":907189,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-06 10:01:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-03-06 08:05:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-02-09 15:02:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":2,"date":"2024-03-01 03:07:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2024-03-06 08:05:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-30 11:29:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-08 15:46:22.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-02-08 20:49:59.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2024-02-09 10:46:52.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"e26008066b99b1a942a7d5857b0616f929f7409f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-01-30 10:05:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b41040bdf0c518432db74ac8b6447a3799945fe7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 11:29:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30524fab9b254c7aa61d23f920cd535d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17dbcfdc763b4fd691df9cbace256e18 : SUCCESS in 9m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/607193fbc79246d0926a5a2e5365c24d : SUCCESS in 8m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ead7700e9b194584826eb587204139fa : SUCCESS in 9m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f554e0a73b244135aff2a754491dcaf0 : SUCCESS in 9m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6f909ba87d1541319943a84390bdff6f : SUCCESS in 9m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d511730cb854c288a5f2c95cd72e1ae : SUCCESS in 11m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af24034e67bd4257801eaed4c81fe346 : SUCCESS in 4m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/36c23418674f4f23aea42901c5df1262 : SUCCESS in 2m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f82049eb99c48c5b1b6a201954488cc : SUCCESS in 8m 34s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/42ef18321c9b42079844366f59f48b60 : SUCCESS in 56m 23s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/4de5e31f9ee04ad1b78893cddb20e583 : FAILURE in 11m 04s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/984bf209d869442c8c5f8da40deb8b1d : SUCCESS in 1h 18m 00s\n\nWarning:\n  Comments left for invalid file .pylintrc","accounts_in_message":[],"_revision_number":1},{"id":"fa4f21937e956d972314de0d71d288bc6481524b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-02-08 15:46:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"50fea848250876b27dbb7f3c58f033e8775314c5","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-02-08 20:49:59.000000000","message":"Patch Set 1: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"059d668f0c7cd1d77eefb925ad0ad7bca66b9ac9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-02-08 22:20:38.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"5e6e4a5515e89f467628890beda0d5185e5e02da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-02-09 08:51:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"7a28f6d778fbcccd9ec529ecc3b9b4f80a291859","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-02-09 09:00:09.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"acf2b76c4b8470210932bc03fdc9446717246546","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-02-09 09:02:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4a3bd5998f4658e617d2aeedca653c64bd6681c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-02-09 09:45:46.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"84e2a512b5b6e763172aea232253682201b89fe2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 11:31:40.000000000","message":"Patch Set 3: Verified+1\n\n(2 comments)\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59cfffa7e41b4613b7087e395b3a86f8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18a87907064a434bb4afc96bf7fd1cd6 : SUCCESS in 8m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/beb74c8e35784ba8bfab6c3f43decc0c : SUCCESS in 8m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1cbcefd77e84e95a3002b8faa85b7e1 : SUCCESS in 7m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c168b59f39574a1a9f1b9f5550ee4de8 : SUCCESS in 7m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/64d3cfb29b234c7a81116e4372ff8677 : SUCCESS in 8m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfa6cbe2a8534a298373b6ee089667b6 : SUCCESS in 10m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41204c4bd71145cd905990963e29dbb9 : SUCCESS in 3m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8d06e3d7e1b94b2785d292e4502b4cd5 : SUCCESS in 2m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0577fb966b75465f9e8d5960407f8938 : SUCCESS in 8m 33s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/f6a7a0dfa5304508b2bd8880d236015f : SUCCESS in 1h 38m 13s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/b732fb3593e34f5aafedfd878f41abda : FAILURE in 8m 41s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/621abf34743a4839a0e98d63cd92aa34 : SUCCESS in 1h 28m 02s","accounts_in_message":[],"_revision_number":3},{"id":"064e847c18e4d8bb7c7bfbf305455bea8f288e4f","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-02-09 15:02:34.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"84ff25e01800d756f45bfb7f89a9043de9e8dcc4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-03-01 03:07:29.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c242c8219723d6d5e5bded8afa1e8aa3c8ce1727","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-03-06 08:05:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"914eb289ab0f87a5b6ce5fbea849057c5cc3ad53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 08:05:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7d61ee8101f63e9c38c14006c100790ccaea18ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 10:01:44.000000000","message":"Patch Set 3: Verified+2\n\n(2 comments)\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a070ee15d1f48e0ac4de1feb5da0681\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a409664911847a69b78cde851d482a7 : SUCCESS in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3aa37a69d945439c8006a4f947b7ee4d : SUCCESS in 7m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff98f78bc14a474cab186c51f0c09551 : SUCCESS in 8m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e2635437cb33413786f3315714ef32d5 : SUCCESS in 7m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d81b47331ccf4fea85d4e1cdb6a916df : SUCCESS in 7m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/809d9d35f1284f5d930afbf9c080dbf4 : SUCCESS in 12m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1facb8cdb40547728149a434b01bd99b : SUCCESS in 4m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d3306047ec9444c8a3a0f5d751767e57 : SUCCESS in 2m 31s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/557c581764be4a41a072133b3f12180b : SUCCESS in 1h 27m 20s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/91af374d75ef4306892ce6ab7a561f5d : SUCCESS in 1h 49m 54s","accounts_in_message":[],"_revision_number":3},{"id":"5176553cc854b65fb8b8a05bec52db13e7ea2c9f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 10:01:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"88413da67f13a4a20490df79c299f03acdcd4fd6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 10:02:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fbcbabeb3bd49cb8ef1449d9ad87c17\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a967549634dc4c5c917d66c461aacf6c : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e851e4b63694c27b9b66546696fe314 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8942c58f20684a1e3f5f944ddbf4967d0468534a","revisions":{"8b3f724e2fdd7efb36eba7b6ea9eb38e10af2772":{"kind":"REWORK","_number":1,"created":"2024-01-30 10:05:11.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/89/907189/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/89/907189/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/89/907189/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/89/907189/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/89/907189/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/89/907189/1"}}},"commit":{"parents":[{"commit":"82183b210b8c511449700f0e9bbb7ae6f9957d05","subject":"Drop b1 from min versions in requirements.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/82183b210b8c511449700f0e9bbb7ae6f9957d05"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-01-30 10:03:52.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-01-30 10:03:52.000000000","tz":60},"subject":"Bump hacking","message":"Bump hacking\n\nhacking 3.0.x is too old.\nTry to synchronize pylint ignore and extension list with\nother Networking projects.\n\nChange-Id: Ieaf7e926c64921b9e92cf04370bff8edb2b0b304\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/8b3f724e2fdd7efb36eba7b6ea9eb38e10af2772"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/8b3f724e2fdd7efb36eba7b6ea9eb38e10af2772"}]},"branch":"refs/heads/master"},"d562d5ad85039970e465d413c59a7d84c2ebb9ab":{"kind":"REWORK","_number":2,"created":"2024-02-09 08:51:11.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/89/907189/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/89/907189/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/89/907189/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/89/907189/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/89/907189/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/89/907189/2"}}},"commit":{"parents":[{"commit":"82183b210b8c511449700f0e9bbb7ae6f9957d05","subject":"Drop b1 from min versions in requirements.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/82183b210b8c511449700f0e9bbb7ae6f9957d05"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-01-30 10:03:52.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-02-09 08:48:13.000000000","tz":60},"subject":"Bump hacking","message":"Bump hacking\n\nhacking 3.0.x is too old.\nTry to synchronize pylint ignore and extension list with\nother Networking projects.\n\nChange-Id: Ieaf7e926c64921b9e92cf04370bff8edb2b0b304\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/d562d5ad85039970e465d413c59a7d84c2ebb9ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/d562d5ad85039970e465d413c59a7d84c2ebb9ab"}]},"branch":"refs/heads/master"},"8942c58f20684a1e3f5f944ddbf4967d0468534a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-02-09 09:45:46.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/89/907189/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/89/907189/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/89/907189/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/89/907189/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/89/907189/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/89/907189/3"}}},"commit":{"parents":[{"commit":"82183b210b8c511449700f0e9bbb7ae6f9957d05","subject":"Drop b1 from min versions in requirements.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/82183b210b8c511449700f0e9bbb7ae6f9957d05"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-01-30 10:03:52.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2024-02-09 09:45:46.000000000","tz":0},"subject":"Bump hacking","message":"Bump hacking\n\nhacking 3.0.x is too old.\nTry to synchronize pylint ignore and extension list with\nother Networking projects.\n\nWith new pip the order of packages is not relevant, so the\nrelated comment from requirements.txts is removed, see pip\ndocumentation:\nhttps://pip.pypa.io/en/stable/cli/pip_install/#installation-order\n\nChange-Id: Ieaf7e926c64921b9e92cf04370bff8edb2b0b304\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/8942c58f20684a1e3f5f944ddbf4967d0468534a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/8942c58f20684a1e3f5f944ddbf4967d0468534a"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
