)]}'
{"id":"openstack%2Fnetworking-bagpipe~911520","triplet_id":"openstack%2Fnetworking-bagpipe~unmaintained%2Fwallaby~I90dae34577b48082fd44b9081f61b2a00fe76519","project":"openstack/networking-bagpipe","branch":"unmaintained/wallaby","topic":"create-wallaby","hashtags":[],"change_id":"I90dae34577b48082fd44b9081f61b2a00fe76519","subject":"Update .gitreview for unmaintained/wallaby","status":"ABANDONED","created":"2024-03-06 12:04:48.000000000","updated":"2025-03-20 08:40:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8f1271a461672941eb987a52910e5dec84c496fd","_number":911520,"virtual_id_number":911520,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-03-06 13:31:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-06 13:31:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-04 08:48:37.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2025-03-20 08:40:23.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"a05b0fc6c859acd47401a224ed6141ff6c6589f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2024-03-06 12:04:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5862efdfda26fe55d820d877617622df23cf83d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 13:31:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/88b941988c4f4791970f9b3f0e9134b8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d4c83527ae244a0aac306692cc3aaa3 : FAILURE in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1601b73133ab4252adb4ff6eeb72ee2d : FAILURE in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca709a94a41243758f49a35a5c7aefa8 : FAILURE in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe598089ab9546608e9178efa00973fe : FAILURE in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05dce95a88db44b79ffe8392a3040912 : FAILURE in 6m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1ff05e830c744b592c8b19ec650f17a : FAILURE in 2m 40s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2171dca11e0049f2a4c908a34301f4eb : FAILURE in 5m 12s\n- neutron-tempest-plugin-bgpvpn-bagpipe-wallaby https://zuul.opendev.org/t/openstack/build/cdb68f3bcaf546748e316ad9a1adb70f : SUCCESS in 1h 16m 18s","accounts_in_message":[],"_revision_number":1},{"id":"8f1271a461672941eb987a52910e5dec84c496fd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-20 08:40:23.000000000","message":"Abandoned\n\nunmaintained/wallaby branch of openstack/networking-bagpipe transitioned to End of Life and is about to be deleted. To be able to do that, all open patches need to be abandoned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6489e3fb77818f44753d8689342b8b3224232dc7","revisions":{"6489e3fb77818f44753d8689342b8b3224232dc7":{"kind":"REWORK","_number":1,"created":"2024-03-06 12:04:48.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/20/911520/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/20/911520/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/20/911520/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/20/911520/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/20/911520/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/20/911520/1"}}},"commit":{"parents":[{"commit":"097de21d696c031972e6f05c8d1bbd3caff320ad","subject":"Add required project and follow pyroute2 changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/097de21d696c031972e6f05c8d1bbd3caff320ad"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-06 12:04:45.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-06 12:04:45.000000000","tz":0},"subject":"Update .gitreview for unmaintained/wallaby","message":"Update .gitreview for unmaintained/wallaby\n\nChange-Id: I90dae34577b48082fd44b9081f61b2a00fe76519\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/6489e3fb77818f44753d8689342b8b3224232dc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/6489e3fb77818f44753d8689342b8b3224232dc7"}]},"branch":"refs/heads/unmaintained/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
