)]}'
{"id":"openstack%2Fnetworking-bagpipe~911545","triplet_id":"openstack%2Fnetworking-bagpipe~unmaintained%2Fxena~Icda2f09fec5f477bd70dfd4aff069cc024060717","project":"openstack/networking-bagpipe","branch":"unmaintained/xena","topic":"create-xena","hashtags":[],"change_id":"Icda2f09fec5f477bd70dfd4aff069cc024060717","subject":"Update .gitreview for unmaintained/xena","status":"ABANDONED","created":"2024-03-06 12:14:37.000000000","updated":"2025-03-21 18:34:41.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"63d95a284c9206f28566aef656e3b311e088d6c0","_number":911545,"virtual_id_number":911545,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-03-06 12:33:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-06 12:33:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-06 15:30:15.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2025-03-21 18:34:41.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"26067427672489833228e5a8c5d3b4aedbc7b971","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2024-03-06 12:14:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a47de77d4b47d8500065b2f45d0b8f0f8fff4443","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 12:33:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8fe995ae84d41f0a731fce07a146cca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa7fd7312d2d4f70881809aad751d218 : FAILURE in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44ae20d2984645c7b4d8fd7b3b2aad55 : FAILURE in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1300e4393044ae88c969f9bc1901669 : FAILURE in 6m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53d59e41259742db88577d75f962dce5 : FAILURE in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfad5ecb79254152afe5ee383dddd44e : FAILURE in 6m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8b5f25c38564b50907efbeae71efe77 : FAILURE in 3m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71dbe5b25a4e4b5095ba68d37a695b8f : FAILURE in 5m 18s","accounts_in_message":[],"_revision_number":1},{"id":"e94089ef6945728fd9e049fa90344ce34164d523","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-06 15:30:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fb0b89d5e8666e7797555c13108d84d447eaf582","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 16:31:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b0c88ea79394721b8c6d39669b7c3bd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be77984b00bd46219c94e2affcc3598f : FAILURE in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34e8ecdb70944259bfd08514f7ecceab : FAILURE in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2aae36debe84ad2a152117fe251d7b1 : FAILURE in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4fd28cbb0b549cb8792c4970126796b : FAILURE in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6514e9c05ff44d80ba62e1897148d9e6 : FAILURE in 5m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54114e4e54bc47babfe59e3fe6728a6b : FAILURE in 2m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4299e93bb96e43e2b4cb32e5dda1f5ed : FAILURE in 4m 27s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/f6b7fcf66daf4a70a80942bbe0c107fd : FAILURE in 25m 24s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/93c73154fc8d4e36b7bd5eb2be26b1ae : FAILURE in 8m 41s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/e0a1769a98724bb79f9bd95ea97d3e80 : FAILURE in 52m 33s","accounts_in_message":[],"_revision_number":1},{"id":"63d95a284c9206f28566aef656e3b311e088d6c0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-21 18:34:41.000000000","message":"Abandoned\n\nunmaintained/xena branch of openstack/networking-bagpipe transitioned to End of Life and is about to be deleted. To be able to do that, all open patches need to be abandoned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4963707dc7f08b17bae20cc3cd67e1bf58c8ff0e","revisions":{"4963707dc7f08b17bae20cc3cd67e1bf58c8ff0e":{"kind":"REWORK","_number":1,"created":"2024-03-06 12:14:37.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/45/911545/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/45/911545/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/45/911545/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/45/911545/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/45/911545/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/45/911545/1"}}},"commit":{"parents":[{"commit":"39546f061193de8d46d6ca6141122f83bb18621a","subject":"Add required projects where necessary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/39546f061193de8d46d6ca6141122f83bb18621a"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-06 12:14:35.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-06 12:14:35.000000000","tz":0},"subject":"Update .gitreview for unmaintained/xena","message":"Update .gitreview for unmaintained/xena\n\nChange-Id: Icda2f09fec5f477bd70dfd4aff069cc024060717\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/4963707dc7f08b17bae20cc3cd67e1bf58c8ff0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/4963707dc7f08b17bae20cc3cd67e1bf58c8ff0e"}]},"branch":"refs/heads/unmaintained/xena"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
