)]}'
{"id":"openstack%2Fnetworking-bagpipe~914272","triplet_id":"openstack%2Fnetworking-bagpipe~master~I4221652a0ba256edf83208ff41e2e2ec24aa190f","project":"openstack/networking-bagpipe","branch":"master","topic":"update-testing-runtime","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-05-21 16:59:34.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-03-27 17:38:35.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I4221652a0ba256edf83208ff41e2e2ec24aa190f","subject":"Update jobs based on testing runtime for 2024.2","status":"MERGED","created":"2024-03-26 16:02:01.000000000","updated":"2024-05-21 17:01:50.000000000","submitted":"2024-05-21 16:59:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914272-update-testing-runtime","meta_rev_id":"88a2a0955e19b7269b57fdfe1679711188cf5287","_number":914272,"virtual_id_number":914272,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-21 16:59:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-03-27 10:53:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-05-21 15:29:40.000000000","_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":1,"date":"2024-03-27 17:38:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-05-21 15:29:40.000000000","_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-26 17:47:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-26 17:56:42.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2024-03-27 10:53:05.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-05-21 15:29:40.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"}],"messages":[{"id":"50aec01936142cff862a33802c094df1bff76233","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-03-26 16:02:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa17d5f914c6cb94db07b515707a49f2d8124041","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-03-26 16:06:42.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a9679e643d9b664b2afe3f8cce61742deecb1bf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 17:47:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b929423156a64e4499e04a3a36d28892\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e667f379fc844fdb9f6f345ce7fb2789 : SUCCESS in 8m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5181eced11540a09360e421d4fda455 : SUCCESS in 8m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e40c516db5040dd9d11bdd600a5a790 : SUCCESS in 8m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/32a245cb1dc54fdcb955a9e7eda463b6 : SUCCESS in 10m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fb31cabaf83b47009f32824ba3b6000d : FAILURE in 12m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c21ae64757b54c70a1167456a0e7670d : SUCCESS in 11m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0808dc8d740f473fb2e7cd32fe344d31 : SUCCESS in 7m 49s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/ded52564c96f470cb7e503b35713c394 : SUCCESS in 1h 30m 53s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/78c0f6dfdd234de2a325e4299326703b : FAILURE in 20m 03s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/f63594597e5a4cb1ba48468240a065d4 : SUCCESS in 1h 10m 50s\n\nWarning:\n  Comments left for invalid file .pylintrc","accounts_in_message":[],"_revision_number":2},{"id":"c852a870adb941e091397ca63a41800524541ef9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-03-26 17:56:42.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1d2780a114837ca1de94e467ce415d1b6b21e0db","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-03-26 18:30:44.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ed0db4f2996ba86c660683e62dd5621bb8084dc1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-03-26 19:52:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aeb0c3a0187a58f361579dd155097895c6d2f9da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-03-27 08:42:25.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"4506a94ca6633f763dabf31296485a9446ee1c86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-27 10:25:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fc15e5508554880af5e3c2f2ef8e278\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5f1ec2e888f4c9192cda55546ec6454 : SUCCESS in 7m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/086a26e5080044efaa837896d9c5f602 : SUCCESS in 7m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45a3c2d4968a495b949d33f518c4ef25 : SUCCESS in 8m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ad4f25380e543a9a808d3e5e454376d : SUCCESS in 6m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b99e1b754fc94da7a273c4fdc5943d57 : FAILURE in 12m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ead0fbf2b62b4ef99bdb4d990a0c10aa : SUCCESS in 9m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f4025ba03a6412399e5cae5fbd6fd2d : SUCCESS in 9m 35s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/8d9dd6b33d4b440d9c5f81cf37dab402 : SUCCESS in 1h 36m 40s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/c9327d7038c54e429300c0766a88b610 : FAILURE in 7m 55s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/5dc92dadff2e4506a1d897fad9b21b98 : SUCCESS in 1h 25m 10s\n\nWarning:\n  Comments left for invalid file .pylintrc","accounts_in_message":[],"_revision_number":3},{"id":"563ef9f0601b00dd71a3a3968dbd1fa905636ab8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-03-27 10:53:05.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f6e383c16e6fc9b64bd47763077f9d93dffabc91","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-03-27 17:38:35.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"de59550b43fdbd7223c9404227c64468a1d33146","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-21 15:29:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2e4777841ae128e9df61ebc835b20adf428b18d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 15:30:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"eed18c190057db3c3d4afa680f958b7666979660","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 16:59:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b641ad1ba3746b2bc1851ecfa4af9c9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3de4a88effb48c1815f1f812f677739 : SUCCESS in 9m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8076f01b05e4abe9a2462e5a704120c : SUCCESS in 7m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e01fb1d145514d0a9cd009a4096ff056 : SUCCESS in 9m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/95b1225085b9454c867bfe1cbb8e48a7 : SUCCESS in 9m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11f60527e5e64995869fde59535f7e58 : SUCCESS in 11m 58s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/0608617e8f8542448ecc267d9be8efbd : SUCCESS in 1h 26m 51s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/95c62b95abb94d10bb8938f6823f7cf7 : SUCCESS in 1h 12m 07s\n\nWarning:\n  Comments left for invalid file .pylintrc","accounts_in_message":[],"_revision_number":3},{"id":"0bb7184803334268c8061d17c7c9c7ccb306504a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 16:59:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"88a2a0955e19b7269b57fdfe1679711188cf5287","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 17:01:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2ccf136430c4e56b36f2628365a6cb0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c8732198e744c6587d7fa6e5119c20a : SUCCESS in 51s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"18d7eab28e5680a48759c438239e664b89ecb28c","revisions":{"8585e538919c49c82568f1bb2db6beb2e62e7b86":{"kind":"REWORK","_number":1,"created":"2024-03-26 16:02:01.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/72/914272/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/72/914272/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/914272/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/914272/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/914272/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/914272/1"}}},"commit":{"parents":[{"commit":"c7d034dc874905d80e6640a3b367662fafcd9d29","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c7d034dc874905d80e6640a3b367662fafcd9d29"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-03-26 16:00:15.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-03-26 16:00:15.000000000","tz":60},"subject":"Update jobs based on testing runtime for 2024.2","message":"Update jobs based on testing runtime for 2024.2\n\nBased on [0] \u0026 [1] only py39, py311 and py312 (non-voting)\njobs are in the testing runtime for 2024.2, updated relevant\nfiles accordingly. Added timeout override for py312\nsimilar to other versions.\n\n[0] https://review.opendev.org/c/openstack/governance/+/908862\n[1] https://governance.openstack.org/tc/reference/runtimes/2024.2.html\n\nChange-Id: I4221652a0ba256edf83208ff41e2e2ec24aa190f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/8585e538919c49c82568f1bb2db6beb2e62e7b86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/8585e538919c49c82568f1bb2db6beb2e62e7b86"}]},"branch":"refs/heads/master"},"7f19571c956db861a98754693118c628dc58d351":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-03-26 16:06:42.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/72/914272/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/72/914272/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/914272/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/914272/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/914272/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/914272/2"}}},"commit":{"parents":[{"commit":"c7d034dc874905d80e6640a3b367662fafcd9d29","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c7d034dc874905d80e6640a3b367662fafcd9d29"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-03-26 16:00:15.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-03-26 16:06:20.000000000","tz":60},"subject":"Update jobs based on testing runtime for 2024.2","message":"Update jobs based on testing runtime for 2024.2\n\nBased on [0] \u0026 [1] only py39, py311 and py312 (non-voting)\njobs are in the testing runtime for 2024.2, updated relevant\nfiles accordingly.\n\n[0] https://review.opendev.org/c/openstack/governance/+/908862\n[1] https://governance.openstack.org/tc/reference/runtimes/2024.2.html\n\nChange-Id: I4221652a0ba256edf83208ff41e2e2ec24aa190f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7f19571c956db861a98754693118c628dc58d351"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7f19571c956db861a98754693118c628dc58d351"}]},"branch":"refs/heads/master"},"18d7eab28e5680a48759c438239e664b89ecb28c":{"kind":"REWORK","_number":3,"created":"2024-03-27 08:42:25.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/72/914272/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/72/914272/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/914272/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/914272/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/914272/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/72/914272/3"}}},"commit":{"parents":[{"commit":"c7d034dc874905d80e6640a3b367662fafcd9d29","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c7d034dc874905d80e6640a3b367662fafcd9d29"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-03-26 16:00:15.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-03-27 08:42:01.000000000","tz":60},"subject":"Update jobs based on testing runtime for 2024.2","message":"Update jobs based on testing runtime for 2024.2\n\nBased on [0] \u0026 [1] only py39, py311 and py312 (non-voting)\njobs are in the testing runtime for 2024.2, updated relevant\nfiles accordingly.\n\n[0] https://review.opendev.org/c/openstack/governance/+/908862\n[1] https://governance.openstack.org/tc/reference/runtimes/2024.2.html\n\nChange-Id: I4221652a0ba256edf83208ff41e2e2ec24aa190f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/18d7eab28e5680a48759c438239e664b89ecb28c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/18d7eab28e5680a48759c438239e664b89ecb28c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
