)]}'
{"id":"openstack%2Fnetworking-bagpipe~916718","triplet_id":"openstack%2Fnetworking-bagpipe~master~I1485436da68ab048ea86de6c27e810824a379bcf","project":"openstack/networking-bagpipe","branch":"master","topic":"json-dumps-encoding","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-06-26 22:32:10.000000000","reason":"Change was submitted"},"36866":{"account":{"_account_id":36866,"name":"Valentin Chassignol","display_name":"Vinetos","email":"contact@vinetos.fr","username":"vinetos"},"last_update":"2024-06-26 22:32:10.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-04-26 09:23:51.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-06-26 20:55:12.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I1485436da68ab048ea86de6c27e810824a379bcf","subject":"rest_attach: encode POST body in UTF-8","status":"MERGED","created":"2024-04-23 08:02:46.000000000","updated":"2024-06-26 22:33:05.000000000","submitted":"2024-06-26 22:32:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916718-json-dumps-encoding","meta_rev_id":"001c04a67f1340fbcd752daa30b2aeeb6acbe23c","_number":916718,"virtual_id_number":916718,"owner":{"_account_id":36866,"name":"Valentin Chassignol","display_name":"Vinetos","email":"contact@vinetos.fr","username":"vinetos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-26 22:32:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2024-04-26 09:23:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2024-06-26 20:55:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2024-06-26 20:55:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-23 10:02:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-24 08:41:08.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2024-04-26 09:23:51.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-06-26 12:20:12.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-06-26 12:20:12.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"16a6d7f89a220b7ff15ac6ff9263804e148023d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36866,"name":"Valentin Chassignol","display_name":"Vinetos","email":"contact@vinetos.fr","username":"vinetos"},"date":"2024-04-23 08:02:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ec82cb508d21a57d8a003cdbf3abdaa31ad9501","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 10:02:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1be79e40e51f402db8601a782ba60762\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c3502a157a84c98946152b1a0b308fe : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef4d26ec4a64419fbfb99ae729861db3 : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9241bb7a2a24566aa266d41aa44adf2 : SUCCESS in 7m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/05317710aa554a6c9debdc399bfad291 : SUCCESS in 7m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ca11849ba6bc4ee1a730f59ddb6e4195 : SUCCESS in 9m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d0cfbacba18b437f894dfa5549d82040 : FAILURE in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1de5b63d8fb14bfa8ba990b25e4ddfba : SUCCESS in 11m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/475950cb62044a888c23d3497662cd7d : SUCCESS in 9m 13s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/94047e17df2e496382cece5680679525 : SUCCESS in 1h 29m 10s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/e2093a0a0dd44815aaa9aa2d0107a158 : FAILURE in 17m 47s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/32d7b0d77c994b4ebbd4f349ebe10ea4 : SUCCESS in 1h 53m 22s\n\nWarning:\n  Comments left for invalid file .pylintrc","accounts_in_message":[],"_revision_number":1},{"id":"2987ce18f85953fed27b0742686a8f978c0ab0a1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-04-24 08:41:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"987151f8d7edd73aedf0b8d671015ac9e77d32e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36866,"name":"Valentin Chassignol","display_name":"Vinetos","email":"contact@vinetos.fr","username":"vinetos"},"date":"2024-04-24 11:56:55.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"86124ae291a0b4ce5338c65a815d6e7839638368","author":{"_account_id":36866,"name":"Valentin Chassignol","display_name":"Vinetos","email":"contact@vinetos.fr","username":"vinetos"},"date":"2024-04-24 11:59:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1463e704fc0965f4440690e248efc166ff89b591","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 13:25:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a48c1f631fe4192bcda6bbc28a06e30\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd01cd6e108e4b6eaf8220fa2e8263e0 : SUCCESS in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82916da096974453ad5408e14b84104a : SUCCESS in 7m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8494d68d8f9749579a166df1e2d6f7e0 : SUCCESS in 7m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/117151dca1184dcd98dfc7f224cc7813 : SUCCESS in 6m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4f5d5b8c50514b5085013d75948f07ad : SUCCESS in 7m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/46be1129981541d985702fdff6c787eb : FAILURE in 11m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/882adc1bb0eb4cdd8403b1cf08d5daec : SUCCESS in 10m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/864ef00eb4e342e9a235c90d6320e90b : SUCCESS in 9m 29s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/c05beabdcac24aba818f7985d1897272 : SUCCESS in 1h 25m 13s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/78dbdb80eff24303a9ebe68ee12f8c37 : FAILURE in 18m 02s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/7c418d625bbf4124bfa0691268383766 : SUCCESS in 1h 25m 49s\n\nWarning:\n  Comments left for invalid file .pylintrc","accounts_in_message":[],"_revision_number":2},{"id":"7ab560c681e611512417a7f8404c42f2fc3a61a9","author":{"_account_id":36866,"name":"Valentin Chassignol","display_name":"Vinetos","email":"contact@vinetos.fr","username":"vinetos"},"date":"2024-04-25 11:59:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7b441bbba39a01ddc8b880b9ab80902ca8760372","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-04-26 09:23:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a5f8937223609e7bce39bb3bf921ba02676bb62e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-06-26 20:55:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6ccad5535f808cb0aab1fb93f3c3cf9577948995","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 20:55:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7dcfaa88beb10cbf23db2c57a6e604db18c7133c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 22:32:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4d44268136745179000e61e2f36f782\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2d43d104c24452a83f1446d3c853e05 : SUCCESS in 9m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ed7417c33d6468abfa5c973dd01760e : SUCCESS in 9m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5568ace6ddd8406887d1e5ab87aca6cd : SUCCESS in 9m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d78dd366a0ab41c9bf501f330b8e1ebd : SUCCESS in 8m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73d8b1f6884e48bcbda538db0417dd19 : SUCCESS in 12m 47s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/acd345abff55404396637eea363e6195 : SUCCESS in 1h 35m 12s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/6c9bc77126cc468290230395719be398 : SUCCESS in 1h 31m 45s\n\nWarning:\n  Comments left for invalid file .pylintrc","accounts_in_message":[],"_revision_number":2},{"id":"cd56f23929c368bc62dcba148b5dcccea21097eb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 22:32:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"001c04a67f1340fbcd752daa30b2aeeb6acbe23c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 22:33:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45b3cdb32e8a4d7188d62fb1dd6aac8e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2593e587bc6479095980c18a1b690ef : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1dc56419c72669076cc4a083dc731912c466bbe3","revisions":{"030b712922911fdaa7aef68c0aef5840c33a4050":{"kind":"REWORK","_number":1,"created":"2024-04-23 08:02:46.000000000","uploader":{"_account_id":36866,"name":"Valentin Chassignol","display_name":"Vinetos","email":"contact@vinetos.fr","username":"vinetos"},"ref":"refs/changes/18/916718/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/18/916718/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/18/916718/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/18/916718/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/18/916718/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/18/916718/1"}}},"commit":{"parents":[{"commit":"385b6b4beba5cca55a4a424fdb1c63d12f4e585d","subject":"Merge \"coveragerc: Remove reference to non-existant path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/385b6b4beba5cca55a4a424fdb1c63d12f4e585d"}]}],"author":{"name":"Valentin Chassignol","email":"valentin.chassignol@infomaniak.com","date":"2024-04-23 07:56:40.000000000","tz":120},"committer":{"name":"Valentin Chassignol","email":"valentin.chassignol@infomaniak.com","date":"2024-04-23 07:56:40.000000000","tz":120},"subject":"rest_attach: encode POST body in UTF-8","message":"rest_attach: encode POST body in UTF-8\n\nURLLib require the request POST-body to be\nencoded in order to be sended.\nThis patch encode the request body in UTF-8.\n\nDocImpact\nCloses-Bug: #2063116\n\nChange-Id: I1485436da68ab048ea86de6c27e810824a379bcf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/030b712922911fdaa7aef68c0aef5840c33a4050"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/030b712922911fdaa7aef68c0aef5840c33a4050"}]},"branch":"refs/heads/master"},"1dc56419c72669076cc4a083dc731912c466bbe3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-04-24 11:56:55.000000000","uploader":{"_account_id":36866,"name":"Valentin Chassignol","display_name":"Vinetos","email":"contact@vinetos.fr","username":"vinetos"},"ref":"refs/changes/18/916718/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/18/916718/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/18/916718/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/18/916718/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/18/916718/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/18/916718/2"}}},"commit":{"parents":[{"commit":"385b6b4beba5cca55a4a424fdb1c63d12f4e585d","subject":"Merge \"coveragerc: Remove reference to non-existant path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/385b6b4beba5cca55a4a424fdb1c63d12f4e585d"}]}],"author":{"name":"Valentin Chassignol","email":"valentin.chassignol@infomaniak.com","date":"2024-04-23 07:56:40.000000000","tz":120},"committer":{"name":"Valentin Chassignol","email":"valentin.chassignol@infomaniak.com","date":"2024-04-24 11:56:55.000000000","tz":0},"subject":"rest_attach: encode POST body in UTF-8","message":"rest_attach: encode POST body in UTF-8\n\nURLLib require the request POST-body to be\nencoded in order to be sended.\nThis patch encode the request body in UTF-8.\n\nCloses-Bug: #2063116\n\nChange-Id: I1485436da68ab048ea86de6c27e810824a379bcf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/1dc56419c72669076cc4a083dc731912c466bbe3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/1dc56419c72669076cc4a083dc731912c466bbe3"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
