)]}'
{"id":"openstack%2Fnetworking-bagpipe~917545","triplet_id":"openstack%2Fnetworking-bagpipe~unmaintained%2Fzed~I1bfa5b70ec8d6662979642f5fdd0fe81af1701f3","project":"openstack/networking-bagpipe","branch":"unmaintained/zed","topic":"create-zed","hashtags":[],"change_id":"I1bfa5b70ec8d6662979642f5fdd0fe81af1701f3","subject":"Update .gitreview for unmaintained/zed","status":"MERGED","created":"2024-04-30 08:51:37.000000000","updated":"2024-06-12 13:12:56.000000000","submitted":"2024-06-12 13:11:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917545-create-zed","meta_rev_id":"40efd8b463058cf337ec80648aa49e93377c2da8","_number":917545,"virtual_id_number":917545,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-12 13:11:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2024-06-12 11:27:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2024-06-12 11:27:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-30 09:19:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-12 08:59:56.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2024-06-12 11:27:49.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"af2049150806107bb3f5eb08e83b21c946930070","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2024-04-30 08:51:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b33fb52995ba32d5db62e27e31d46d79623dcec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 09:19:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/db07b7a289054e7fb7a3225f14dd296a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f26657df73e4bce88f905ca2bf60c68 : FAILURE in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78a9c833099d4a88a955e2cef5255bdd : FAILURE in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d764ed3a89fd468c82fe3353b76367f4 : FAILURE in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/be9f7a393fb94cacb8916e7f7cf3e96c : FAILURE in 7m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/517ee4fe40f94d83a5b63f2951414171 : FAILURE in 5m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9e14469831f4b3d88d26ff2c340cc7e : FAILURE in 5m 12s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/8378a0ab81b14a25a8e36aa13ed5f888 : FAILURE in 20m 50s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/0c4ba11e1be847fe90c47eae98ae1e5b : FAILURE in 18m 22s","accounts_in_message":[],"_revision_number":1},{"id":"0e4a3bed3fbe4783ee081f1cf653ce003b300fa6","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-06-12 08:59:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"444cd2d46fe05829429110e53d76cf48f4719073","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 10:59:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d69dbb51e1bd4b48905e8b6744671428\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/884300578ba94487928de800271e46d0 : SUCCESS in 8m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11afdcbf9e1e467fa321c61d6622585e : SUCCESS in 8m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bde7f7fb4bb40ecb620c0fe5c6c826a : SUCCESS in 9m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2fd3fce60e8d4a06b499018f8a851c86 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f86f157c33a9437382309474aa811598 : SUCCESS in 9m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d7960b6e0ab422ba374adee8bb2c437 : SUCCESS in 9m 17s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/974748c19ddb4f0980512a0137fd2ee9 : SUCCESS in 1h 54m 05s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/2d14d38cf7f64f27b357b0890d917cd3 : FAILURE in 16m 40s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/73cc5dca4aa449a1b0ecbfd536d03c46 : SUCCESS in 1h 12m 32s","accounts_in_message":[],"_revision_number":1},{"id":"c7fcb8ded407f2e66376743fc20c36e23732a323","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-06-12 11:27:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e128e6354d7f067e6a5e51c75832c6c138c23adf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 11:28:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1397ba38438d632a6950167836c85151eec56352","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 13:11:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d71ba2f8572446db9388095438ed256\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cd9ac7a9eb04b82a09c28919ddf5ce0 : SUCCESS in 9m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a77696e79a9644d397db9dbb407e694f : SUCCESS in 7m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e235d3ffda34d51b480cb292922f5ba : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95e64b721d3e453aaa7dfd3a82d8a2c5 : SUCCESS in 12m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c77518418d404d5ca8540fead28c16e2 : SUCCESS in 8m 08s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/a034eff5058044a5913e70b5c6920623 : SUCCESS in 1h 38m 10s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/1fb4638d725e4f1aa94eaf015d4795ab : SUCCESS in 1h 13m 36s","accounts_in_message":[],"_revision_number":1},{"id":"a34c23282acfae48b4455ccc4c5f8f48e2aec12a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 13:11:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"40efd8b463058cf337ec80648aa49e93377c2da8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 13:12:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4181a49dfe8e4c068c3bd80e471fb307\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb8b4ce2e3c14264ae4d4cddafac499b : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c84078343e791454f7f6f622138cc19aec7e7be","revisions":{"6c84078343e791454f7f6f622138cc19aec7e7be":{"kind":"REWORK","_number":1,"created":"2024-04-30 08:51:37.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/45/917545/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/45/917545/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/45/917545/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/45/917545/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/45/917545/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/45/917545/1"}}},"commit":{"parents":[{"commit":"1b474c3333b25970431ade31d90a38af3f47fb7a","subject":"Add required projects where necessary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/1b474c3333b25970431ade31d90a38af3f47fb7a"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-04-30 08:51:36.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-04-30 08:51:36.000000000","tz":0},"subject":"Update .gitreview for unmaintained/zed","message":"Update .gitreview for unmaintained/zed\n\nChange-Id: I1bfa5b70ec8d6662979642f5fdd0fe81af1701f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/6c84078343e791454f7f6f622138cc19aec7e7be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/6c84078343e791454f7f6f622138cc19aec7e7be"}]},"branch":"refs/heads/unmaintained/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
