)]}'
{"id":"openstack%2Fnetworking-bagpipe~920050","triplet_id":"openstack%2Fnetworking-bagpipe~master~Ic7385be4cd3e2e02f743144f20734cfe79c57d21","project":"openstack/networking-bagpipe","branch":"master","topic":"fix_str_url","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-05-21 08:17:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic7385be4cd3e2e02f743144f20734cfe79c57d21","subject":"do not use str(url) to stringify a URL for subsequent use","status":"MERGED","created":"2024-05-20 15:35:32.000000000","updated":"2024-05-21 08:19:07.000000000","submitted":"2024-05-21 08:17:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"920050-fix_str_url","meta_rev_id":"81d2b6f81b5eb7c1a487ebcf068ae4e7b06bde91","_number":920050,"virtual_id_number":920050,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-21 08:17:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2024-05-21 06:33:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2024-05-20 19:49:14.000000000","_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2024-05-21 06:33:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-20 16:59:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-20 19:49:14.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2024-05-21 06:33:55.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"1ce12053c0b5f4d0ffed988818f17b3fd556f685","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-20 15:35:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3107a37dcb580b481e475cfa29a9d7024dabca07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 16:59:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3636bd76e6c24b6d9324c3bede224ad8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba57e905616d4ce1877bff7ef87f124b : SUCCESS in 9m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca69c4cb14c44188bd61d965a0cb5bc3 : SUCCESS in 7m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15af182843a14e90a5796a1aed47a817 : SUCCESS in 9m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2115ada6efc14bfc9add80a3fb0ac787 : SUCCESS in 9m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/21badfacaea645498d19be99d9a3106e : SUCCESS in 8m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7f4a2196868744159c28a81854ccd6e3 : FAILURE in 12m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/630acf9763454f84b170735100c9b8e1 : SUCCESS in 12m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e87e47021e44aaab04709104a55fbf7 : SUCCESS in 8m 55s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/c67e404bd7ca41a281d3be61ba0d4e3a : SUCCESS in 1h 22m 32s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/8102ee42f8894980adbcf399a6cca19d : FAILURE in 16m 29s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/02d31b59469340c784a1d64b8461c30e : SUCCESS in 1h 13m 54s\n\nWarning:\n  Comments left for invalid file .pylintrc","accounts_in_message":[],"_revision_number":1},{"id":"6289eb6277467e2836796a7254ce01898262fe4d","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-20 19:49:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"247a2c973d8e32790e3a3d406085bac4eaedc293","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-05-21 06:33:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fb92e3a03d5860c58ee53a3d3f58440fb807aa37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 06:34:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bdb1488a18a20efcf2aa01676c13f884f743b0bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 08:17:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0dbdc49a84664529b4a8590c588c461c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a2e175208a9439c9276820041bca3e4 : SUCCESS in 10m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54ebbebf680d467f9c801834855fac7b : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b36e112c2c2d45bda15b355a712fe7ac : SUCCESS in 9m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d05a374c59034e57973dda408b2f24ab : SUCCESS in 7m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8b204da80cb74c58b01e9d67e27b62df : SUCCESS in 7m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dc0e95131aa485e880c43629da3ef16 : SUCCESS in 11m 18s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/2c2b20173dbd42d89c8627e701ad0b5a : SUCCESS in 1h 28m 22s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/f32f21c894404f1c88c1593d0ef6e395 : SUCCESS in 1h 37m 32s\n\nWarning:\n  Comments left for invalid file .pylintrc","accounts_in_message":[],"_revision_number":1},{"id":"63a576870efa36463b0f24d1965af7674afb1b3f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 08:17:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"81d2b6f81b5eb7c1a487ebcf068ae4e7b06bde91","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 08:19:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a579976cc9624d438a062f44fa8ea78d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38669e58446d4b3a966c402119fd5548 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b12ef2da50d8dc5ddf1c71cf55035a0444ef6c9","revisions":{"1b12ef2da50d8dc5ddf1c71cf55035a0444ef6c9":{"kind":"REWORK","_number":1,"created":"2024-05-20 15:35:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/920050/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/50/920050/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/50/920050/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/50/920050/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/50/920050/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/50/920050/1"}}},"commit":{"parents":[{"commit":"385b6b4beba5cca55a4a424fdb1c63d12f4e585d","subject":"Merge \"coveragerc: Remove reference to non-existant path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/385b6b4beba5cca55a4a424fdb1c63d12f4e585d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-20 15:35:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-20 15:35:15.000000000","tz":540},"subject":"do not use str(url) to stringify a URL for subsequent use","message":"do not use str(url) to stringify a URL for subsequent use\n\nThe str(url) function in SQLAlchemy hides the password.\nFor a URL string that is to be re-used, use\nrender_as_string(hide_password\u003dFalse).\n\nCo-Authored-By: Mike Bayer \u003cmike_mp@zzzcomputing.com\u003e\nChange-Id: Ic7385be4cd3e2e02f743144f20734cfe79c57d21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/1b12ef2da50d8dc5ddf1c71cf55035a0444ef6c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/1b12ef2da50d8dc5ddf1c71cf55035a0444ef6c9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
