)]}'
{"id":"openstack%2Fnetworking-bagpipe~930818","triplet_id":"openstack%2Fnetworking-bagpipe~master~Ib7148a6c7983e54897effe613f780114c062fa1d","project":"openstack/networking-bagpipe","branch":"master","hashtags":[],"change_id":"Ib7148a6c7983e54897effe613f780114c062fa1d","subject":"More strictly validate [sfc_bagpipe] rtnn option","status":"ABANDONED","created":"2024-09-29 04:30:28.000000000","updated":"2025-07-20 10:25:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"53794880270ab5034547310d74740ec59a85bde6","_number":930818,"virtual_id_number":930818,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-09-29 07:06:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-29 07:06:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f8882f8be51401c2b9eaa2cbb6f368c3752ea873","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-29 04:30:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1cb88ee6bc39714b3897e793a8cda9e4e11dcefc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-29 05:25:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9383f1e058b124f267e82137d15ae5121bc443bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-29 07:06:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c03a9f09de5c402f87285d5d6b4a5112\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71eb4e22b7de44cc9497a165e455b488 : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80bc0d5d902e40cfb52d616176fa6554 : SUCCESS in 5m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b73b6d37b7844f2787784da9d117b8ac : SUCCESS in 5m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/06b86f651c9647df9d4c0ca79e26e58f : SUCCESS in 5m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fa7ba10e514d46439dbf6444b1d14960 : SUCCESS in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2d3d38ed33e4be7bd1401bbe26021fb : SUCCESS in 11m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80513bdf5d6e4e429c4a73653d0281fd : SUCCESS in 7m 14s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/1e780973030c4b74bb29d82ea03ef176 : SUCCESS in 1h 39m 58s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/ea9cdc4706844b20a510cdc7c394c7f7 : FAILURE in 9m 24s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/1a9bee66482748059cd35c53f869c7bb : SUCCESS in 1h 30m 06s\n\nWarning:\n  Comments left for invalid file .pylintrc","accounts_in_message":[],"_revision_number":2},{"id":"53794880270ab5034547310d74740ec59a85bde6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-20 10:25:35.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a3ec2f28d903e912fca9d873c2c2d7325a082a3d","revisions":{"960810c4a60bfd88c0516691cb6e675aa7152b73":{"kind":"REWORK","_number":1,"created":"2024-09-29 04:30:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/930818/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/18/930818/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/18/930818/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/18/930818/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/18/930818/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/18/930818/1"}}},"commit":{"parents":[{"commit":"0cfbb63aa558117d2dc1b5b2e6b1cfb3fa89dfb7","subject":"Treat items of [sfc_bagpipe] rtnn as Integer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0cfbb63aa558117d2dc1b5b2e6b1cfb3fa89dfb7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-29 04:24:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-29 04:29:48.000000000","tz":540},"subject":"More strictly validate [sfc_bagpipe] rtnn option","message":"More strictly validate [sfc_bagpipe] rtnn option\n\nThis option represents a range of BGP Route Targets. Ensure its length\nand item order are appropriate.\n\nChange-Id: Ib7148a6c7983e54897effe613f780114c062fa1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/960810c4a60bfd88c0516691cb6e675aa7152b73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/960810c4a60bfd88c0516691cb6e675aa7152b73"}]},"branch":"refs/heads/master"},"a3ec2f28d903e912fca9d873c2c2d7325a082a3d":{"kind":"REWORK","_number":2,"created":"2024-09-29 05:25:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/930818/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/18/930818/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/18/930818/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/18/930818/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/18/930818/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/18/930818/2"}}},"commit":{"parents":[{"commit":"0cfbb63aa558117d2dc1b5b2e6b1cfb3fa89dfb7","subject":"Treat items of [sfc_bagpipe] rtnn as Integer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0cfbb63aa558117d2dc1b5b2e6b1cfb3fa89dfb7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-29 04:24:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-29 05:25:23.000000000","tz":540},"subject":"More strictly validate [sfc_bagpipe] rtnn option","message":"More strictly validate [sfc_bagpipe] rtnn option\n\nThis option represents a range of BGP Route Targets. Ensure its length\nand item order are appropriate.\n\nChange-Id: Ib7148a6c7983e54897effe613f780114c062fa1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a3ec2f28d903e912fca9d873c2c2d7325a082a3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/a3ec2f28d903e912fca9d873c2c2d7325a082a3d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
