)]}'
{"id":"openstack%2Fnetworking-bagpipe~937014","triplet_id":"openstack%2Fnetworking-bagpipe~master~Ibf2028c49313fbfe857e34bc4a59fb2a5165b592","project":"openstack/networking-bagpipe","branch":"master","topic":"dnm","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-12-12 16:23:21.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ibf2028c49313fbfe857e34bc4a59fb2a5165b592","subject":"DNM: test master","status":"ABANDONED","created":"2024-12-04 13:23:41.000000000","updated":"2024-12-12 16:23:21.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"507ad51a18d82fc0757871312dc40e80713e2825","_number":937014,"virtual_id_number":937014,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-12-04 15:06:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-04 15:06:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"47ffed14e058960d8fb7f215cff2615ac9602ecb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-12-04 13:23:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6211f0c98316aefac8391962cc7683d34e27360b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 15:06:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4d149a7af9343649f39640abce258cb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2d7727d155145a49ca06a9d844864e9 : FAILURE in 12m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5fdfdd750114ccc8f4b9f7da8db5933 : SUCCESS in 8m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fd6d858d08a544528dc5aed1c4b8fde5 : SUCCESS in 12m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/033de965e2d94f7fb6af125d9d346a16 : SUCCESS in 8m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e12225a9bc74293b44e7281c2e470e6 : SUCCESS in 8m 20s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/a21b746495e54f91b70f89030944d99e : SUCCESS in 1h 41m 06s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/7c5f13e243ce456fa06812cc0adc0250 : FAILURE in 7m 50s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/aa4ea490308f43af8148f2762d2d71bb : SUCCESS in 1h 31m 19s\n\nWarning:\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/vpn/evpn/ovs.py\n  Comments left for invalid file networking_bagpipe/db/sfc_db.py\n  Comments left for invalid file networking_bagpipe/driver/sfc.py\n  Comments left for invalid file networking_bagpipe/objects/bgpvpn.py\n  Comments left for invalid file networking_bagpipe/objects/sfc.py","accounts_in_message":[],"_revision_number":1},{"id":"745ca6a94d5a24d65839a761b849338d471a9e4e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-12-06 12:31:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6c4cbe76e5c008deff130f32c1df9bd30a704006","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 14:20:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4fa44e6bf45b433593d52ebc827725af\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5472a73fd974db2a7cbde0ebc65b38b : SUCCESS in 10m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80202f3f251a49e69f149ba142f7e4dc : SUCCESS in 6m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ce8480a842d04adfa18a6f592c32aeb5 : SUCCESS in 10m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3df695bb429a4faf820b4a5c5dd11542 : SUCCESS in 9m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30589dca85954a1da759684487286127 : SUCCESS in 11m 46s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/519455bd4a284a548bdee6e4726e7f07 : SUCCESS in 1h 47m 01s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/6680212eb8dd4a0a9cf8f2ee5b23433d : FAILURE in 9m 16s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/85ae50f436244ec69a01bc0499d1d031 : FAILURE in 1h 22m 32s","accounts_in_message":[],"_revision_number":1},{"id":"507ad51a18d82fc0757871312dc40e80713e2825","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-12-12 16:23:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c874cb4bbab3c07581d9bf7fb61f9275454b250a","revisions":{"c874cb4bbab3c07581d9bf7fb61f9275454b250a":{"kind":"REWORK","_number":1,"created":"2024-12-04 13:23:41.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/14/937014/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/14/937014/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/14/937014/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/14/937014/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/14/937014/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/14/937014/1"}}},"commit":{"parents":[{"commit":"18f38255282e6d112459842c99ac82e5c9dd02f2","subject":"tests: test_db_base_plugin_v2.py was moved to common","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/18f38255282e6d112459842c99ac82e5c9dd02f2"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-12-04 13:23:18.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-12-04 13:23:18.000000000","tz":60},"subject":"DNM: test master","message":"DNM: test master\n\nChange-Id: Ibf2028c49313fbfe857e34bc4a59fb2a5165b592\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c874cb4bbab3c07581d9bf7fb61f9275454b250a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c874cb4bbab3c07581d9bf7fb61f9275454b250a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
