)]}'
{"id":"openstack%2Fnetworking-bagpipe~937102","triplet_id":"openstack%2Fnetworking-bagpipe~master~I09727da2b8806d9019baedcaa7d743075cc58faa","project":"openstack/networking-bagpipe","branch":"master","topic":"pylint_issue","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-12-05 17:24:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I09727da2b8806d9019baedcaa7d743075cc58faa","subject":"pep8: Issue on Noble (U. 24.04) with pylint","status":"MERGED","created":"2024-12-05 10:14:46.000000000","updated":"2024-12-05 17:26:29.000000000","submitted":"2024-12-05 17:24:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937102-pylint_issue","meta_rev_id":"926c5313f54c066987956338cd94641d9579af22","_number":937102,"virtual_id_number":937102,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-05 17:24:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-12-05 15:39:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2024-12-05 15:39:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-05 12:10:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-05 15:39:46.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"25c7f3967fde6983ec44dbbc5f1572cd897ab383","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-12-05 10:14:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a49cca586116ff2aeb1923628fa9c89fd97d9d77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 12:10:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/190beb0d4a014cd583bbd3c3882d0eae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f82ffe95602d45ed9717c4bebdc03f72 : SUCCESS in 10m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14ff29e7c9d445e48398889e45396071 : SUCCESS in 4m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d5cabb7c20994abfa25b2b8fb9451a49 : SUCCESS in 11m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc31d043188644e2b3547e0a7ad8bf06 : SUCCESS in 9m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6331f86225054b31bab4efc726e5c3eb : SUCCESS in 2m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/56c4585d7c7f4a1786d8762d7c59ef2b : SUCCESS in 2m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86b8d77740a04519a8d0e1dd2da63063 : SUCCESS in 12m 02s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/21f92d593e5246c6b23b9053a58b307d : SUCCESS in 1h 50m 25s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/790290b04166485687c443a42cb59b59 : RETRY_LIMIT in 8m 19s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/778c86fb362c499cb3b5b9248691c82a : SUCCESS in 1h 26m 36s","accounts_in_message":[],"_revision_number":1},{"id":"44d6ddb1f986b98da3c17ca775f0dc5d2c8aaf66","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-05 15:39:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"47212f0baebe4ff00e7a5d92088997776511e4b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 15:40:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a576c20ab9cf45a09e8a4eb5f3f8c55c030c25c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 17:24:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9905d028495e4c9a86e1b043fd352bfb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c449e442caf9414db5e0d5f1933917a5 : SUCCESS in 10m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8da63ea4f5e4ca1abc57295d56c6778 : SUCCESS in 7m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ce289aff356b4190bb9faa7ba5a6f5aa : SUCCESS in 9m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25d40bb6269b461d98eb4bc34a42e850 : SUCCESS in 9m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46bf2b18fea647818ba4c9dfdf23cde7 : SUCCESS in 4m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b155a902ea6f4a2b91b3329e0f724562 : SUCCESS in 2m 41s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/0ce11df327b848859782d7d9f57d4728 : SUCCESS in 1h 41m 41s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/3dfe2890d62b4d93803f7bb06e0d92b3 : SUCCESS in 1h 06m 27s","accounts_in_message":[],"_revision_number":1},{"id":"94db632587a9c80c5d7bd84e751ca8102bc83f04","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 17:24:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"926c5313f54c066987956338cd94641d9579af22","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 17:26:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd537555dcdf406aaa344e93e7d393f2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ef0be0c7aff471d9109d9a5f677f26e : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a31765a47364ca6951910f3eccda9d0 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0b6e789830a3fbe3d842d7294823f791dd75b719","revisions":{"0b6e789830a3fbe3d842d7294823f791dd75b719":{"kind":"REWORK","_number":1,"created":"2024-12-05 10:14:46.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/02/937102/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/02/937102/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/02/937102/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/02/937102/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/02/937102/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/02/937102/1"}}},"commit":{"parents":[{"commit":"18f38255282e6d112459842c99ac82e5c9dd02f2","subject":"tests: test_db_base_plugin_v2.py was moved to common","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/18f38255282e6d112459842c99ac82e5c9dd02f2"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-12-05 10:14:20.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-12-05 10:14:20.000000000","tz":60},"subject":"pep8: Issue on Noble (U. 24.04) with pylint","message":"pep8: Issue on Noble (U. 24.04) with pylint\n\nUse pylint 3.2.0 on Noble.\n\nChange-Id: I09727da2b8806d9019baedcaa7d743075cc58faa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0b6e789830a3fbe3d842d7294823f791dd75b719"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0b6e789830a3fbe3d842d7294823f791dd75b719"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
