)]}'
{"id":"openstack%2Fnetworking-bagpipe~940765","triplet_id":"openstack%2Fnetworking-bagpipe~master~I8c4f857f560841a14ad86538279366bab5c9f2ac","project":"openstack/networking-bagpipe","branch":"master","topic":"remove-linuxbridge-agent","attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-03-02 05:01:17.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-02-18 08:45:39.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"7730":{"account":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"last_update":"2025-02-18 22:51:33.000000000","reason":"Change was submitted"},"12021":{"account":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"last_update":"2025-02-18 22:51:33.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-02-17 14:22:37.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-02-18 21:26:26.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I8c4f857f560841a14ad86538279366bab5c9f2ac","subject":"Remove LinuxBridge related code","status":"MERGED","created":"2025-02-05 08:54:47.000000000","updated":"2025-03-02 05:01:17.000000000","submitted":"2025-02-18 22:51:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"940765-remove-linuxbridge-agent","meta_rev_id":"04110fac8ed20358a3f6a7a61c75f5804c0b7e09","_number":940765,"virtual_id_number":940765,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-18 22:51:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-02-18 08:45:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":2,"date":"2025-02-18 21:26:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":1,"date":"2025-02-18 21:26:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"}],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-05 10:33:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-14 15:47:56.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-02-14 17:01:11.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-02-15 16:35:11.000000000","updated_by":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"reviewer":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"state":"REVIEWER"},{"updated":"2025-02-17 08:15:12.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2025-03-02 05:01:17.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"e03ae2630fc95a2b7eb2e9a954bf58c04b9b1f24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-02-05 08:54:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ff29b7a21b3837a7716486ace3fc13ba6aa18c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 10:33:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/564603b78add48d986359c2c8606c4c6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c2c3731277d34550b4687178aa497ff4 : SUCCESS in 2m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14e0aba2defa4fae823b6c5942530663 : SUCCESS in 8m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2aa97d3e878b4016b28367b9eb6b3078 : SUCCESS in 10m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3b826c61cc6744b386a72c13d14facc1 : SUCCESS in 8m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd34435e6c944d76b0e11669fdb29b03 : SUCCESS in 8m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b14a9af477134580aa4b29fa28d5d853 : SUCCESS in 12m 25s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/232f34c9b6e94bf191d999a91a16ae60 : FAILURE in 1h 34m 07s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/4401ecaa94fa40d2a6299668aae1feb2 : FAILURE in 9m 15s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/e03eef6d255b4770afd61b9bed405dfc : SUCCESS in 1h 06m 08s","accounts_in_message":[],"_revision_number":1},{"id":"625dbf43e09f499fefd434f77b0d916ff06afba4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-02-05 12:48:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dcc565488e29e668a094fc8eefdd2b8ee53f0f4a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-02-11 15:53:54.000000000","message":"Topic bug/2097367 removed","accounts_in_message":[],"_revision_number":1},{"id":"51e0c706de80e5ea8563a95b3925380a15a2fc34","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-02-11 15:54:00.000000000","message":"Topic set to remove-linuxbridge-agent","accounts_in_message":[],"_revision_number":1},{"id":"5c009bfcb629ca2691d38d4d9e5a4b17c2fe0f98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-02-14 11:15:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c58ed1dfb973f755a9d4c13154989bfc40cf1bb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 12:58:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af0e5064b74d432c960aa6f9e16c1e8c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5c497d90ac634ad28f4339488153d194 : SUCCESS in 2m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19791ad5edd64f15931adddba952361e : SUCCESS in 8m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/205e473f89894c73ab808c1a4ef58996 : SUCCESS in 6m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8c86eef8580d49f79bc08fe96ef0b681 : SUCCESS in 10m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a7a1beeb2f04726965c974ed04ae243 : SUCCESS in 11m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f09f9d2cc66468ebaff315a6a625397 : SUCCESS in 12m 04s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/8279c5b9a6fc4326adce81a88c9bd975 : SUCCESS in 1h 40m 08s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/b6f3741a65ef41ad98ebde084f6d8b98 : FAILURE in 10m 07s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/ada138abce1d42f6bd8cc63044fc6c38 : SUCCESS in 1h 23m 13s","accounts_in_message":[],"_revision_number":2},{"id":"e5de4dda16b1b5853addeb0bbe76b9a3d1206818","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-02-14 15:47:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5f8d2b521adcb340cbc491f038e32f6952054583","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-02-14 17:01:11.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bb9f7b781e0cc83b5859d81cf69d391e7b96d710","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-15 14:44:41.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a24a506b7b06637fed9016f47b27433e28d879e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 16:09:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/436a8521806a437295f3a631bc0f1270\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/877728208ee749d9922cfac8e44ed3e4 : SUCCESS in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/105f7b9ab7cd4af8bce5ce8709336007 : SUCCESS in 9m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf83dfbe4a3e492e9cfad08b21eb25f5 : SUCCESS in 6m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f5705877303d4e95a6c7d260758ad1d7 : SUCCESS in 10m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab378889e27d419b834771818c570d4b : SUCCESS in 10m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc375853fc044bbabd523c44f7617268 : SUCCESS in 10m 25s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/75d9e1d58ab14db48f7699e769132234 : SUCCESS in 1h 24m 09s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/ba09a36f61134cdaa519d6224249f825 : FAILURE in 10m 00s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/8c05d8bc1c4c48aa8799b94b562ac141 : SUCCESS in 1h 11m 02s","accounts_in_message":[],"_revision_number":3},{"id":"9ff2659b28b124b73de8039bf7557fa0fcaa1e41","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-15 16:35:11.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a5838892f76593a74743857b3eced000f3d372ca","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2025-02-15 16:35:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7b75667f20ff83af669e358fe9f1f6b27a840ad4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-02-17 13:06:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"305c83a34501154a6f8686586e3811e3b53a5611","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-02-17 14:21:37.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"e53e94c58b04a3f4493211bfba2a37824f26db82","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-02-17 14:22:37.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"553c583b2d4601faaf1ebd5b1fcd4fe3acce3e93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 17:10:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91e58623d7aa42f4a80440d01cd1f447\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6bf8fabac5394b4f8665a76d2ae4c23d : SUCCESS in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/751fc1a9005948a1bdd36c4e110bf9ac : SUCCESS in 10m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3caff66478364c7f99dcee6a32770d14 : SUCCESS in 8m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1a92daaf75d64d2a9b54cdb178fff382 : SUCCESS in 19m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01f1db39fc1b443391127c6e8a15a1b2 : SUCCESS in 9m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c63bd9cb04df40f49a4ef604b70dcf11 : SUCCESS in 2m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37bb603f9ba74fa0a9e2a9b5cfb0e2d9 : SUCCESS in 10m 51s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/7be22aeb5f524fab82d6ab2a1c8d59db : SUCCESS in 1h 44m 41s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/547704f61dac4a65a7a2653df91647c0 : FAILURE in 10m 24s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/0c51df4a31c94a66bfbc9411194b5d34 : SUCCESS in 1h 29m 52s","accounts_in_message":[],"_revision_number":4},{"id":"43751266111d657a483a0ad12278bf066103c29e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-02-18 08:45:39.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3de62894fe852d92d40ba8afdba26e4c3cfa6b29","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-02-18 21:26:26.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"66b3da9896fee2e8b83ec2044f95347a5362b50a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 21:27:02.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"442252e3c9192805deea883c3b107e908f5e43f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 22:51:33.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3e9abb0d45c4c1483d657bfd4241690\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6ef4069bdc12479fa2d97f5d50f295ac : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/697f88232817420fbf84c9df05692225 : SUCCESS in 10m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f198f75964c64f7bb84928ee5c34fcba : SUCCESS in 7m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/74be29a6e1f84aedb21c30abdb1ea19d : SUCCESS in 9m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27d83974c61a43ab84e5298ac4dcbb6b : SUCCESS in 10m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5248bb9999d843c8994c76670f2c4086 : SUCCESS in 3m 11s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/18c3026e6352442ba8b28cf5422d4031 : SUCCESS in 1h 23m 40s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/35b3b252cb2c4d159a265779ad491389 : SUCCESS in 1h 23m 01s","accounts_in_message":[],"_revision_number":4},{"id":"7aece64c54047552fdc10f22cf87232ff9e6ad6f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 22:51:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"e5e6919b7f96088530df7ff153c4f9d950d0b07f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 22:52:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8b4997ee04e4cbbb888a58cf1dccc0e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a20e86160d44c9197d48818d2ad233d : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e48723d46e5c414b9b119ecc0a2c7de9 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":4},{"id":"04110fac8ed20358a3f6a7a61c75f5804c0b7e09","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-02 05:01:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ae61d585b651290a8786438058e5c8a1ecdb2063","revisions":{"b9e577651788898010769e2321fe73c20acd9982":{"kind":"REWORK","_number":1,"created":"2025-02-05 08:54:47.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/65/940765/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/65/940765/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/65/940765/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/65/940765/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/65/940765/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/65/940765/1"}}},"commit":{"parents":[{"commit":"0b6e789830a3fbe3d842d7294823f791dd75b719","subject":"pep8: Issue on Noble (U. 24.04) with pylint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0b6e789830a3fbe3d842d7294823f791dd75b719"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-02-05 08:51:05.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-02-05 08:51:05.000000000","tz":60},"subject":"WIP: Remove LinuxBridge related code","message":"WIP: Remove LinuxBridge related code\n\nWith [1] the LinuxBridge driver related code was cut from\nNeutron, let\u0027s try to do the same for networking-bagpipe.\n\n[1]:https://review.opendev.org/c/openstack/neutron/+/927216\nCloses-Bug: #2097367\n\nChange-Id: I8c4f857f560841a14ad86538279366bab5c9f2ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b9e577651788898010769e2321fe73c20acd9982"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b9e577651788898010769e2321fe73c20acd9982"}]},"branch":"refs/heads/master"},"5d67b4e5c2f1a75fb1463da8482d73690c91ab8b":{"kind":"REWORK","_number":2,"created":"2025-02-14 11:15:55.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/65/940765/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/65/940765/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/65/940765/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/65/940765/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/65/940765/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/65/940765/2"}}},"commit":{"parents":[{"commit":"0b6e789830a3fbe3d842d7294823f791dd75b719","subject":"pep8: Issue on Noble (U. 24.04) with pylint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0b6e789830a3fbe3d842d7294823f791dd75b719"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-02-05 08:51:05.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-02-14 11:15:44.000000000","tz":60},"subject":"WIP: Remove LinuxBridge related code","message":"WIP: Remove LinuxBridge related code\n\nWith [1] the LinuxBridge driver related code was cut from\nNeutron, let\u0027s try to do the same for networking-bagpipe.\n\n[1]:https://review.opendev.org/c/openstack/neutron/+/927216\nCloses-Bug: #2097367\n\nChange-Id: I8c4f857f560841a14ad86538279366bab5c9f2ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/5d67b4e5c2f1a75fb1463da8482d73690c91ab8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/5d67b4e5c2f1a75fb1463da8482d73690c91ab8b"}]},"branch":"refs/heads/master"},"7fda9c8810e2ff6dcf50134c063d4611fe7fa865":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-02-15 14:44:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/940765/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/65/940765/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/65/940765/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/65/940765/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/65/940765/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/65/940765/3"}}},"commit":{"parents":[{"commit":"0b6e789830a3fbe3d842d7294823f791dd75b719","subject":"pep8: Issue on Noble (U. 24.04) with pylint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0b6e789830a3fbe3d842d7294823f791dd75b719"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-02-05 08:51:05.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-15 14:44:41.000000000","tz":0},"subject":"Remove LinuxBridge related code","message":"Remove LinuxBridge related code\n\nWith [1] the LinuxBridge driver related code was cut from\nNeutron, let\u0027s try to do the same for networking-bagpipe.\n\n[1]:https://review.opendev.org/c/openstack/neutron/+/927216\nCloses-Bug: #2097367\n\nCloses-Bug: #2098570\nChange-Id: I8c4f857f560841a14ad86538279366bab5c9f2ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7fda9c8810e2ff6dcf50134c063d4611fe7fa865"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7fda9c8810e2ff6dcf50134c063d4611fe7fa865"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ae61d585b651290a8786438058e5c8a1ecdb2063":{"kind":"REWORK","_number":4,"created":"2025-02-17 14:21:37.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/65/940765/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/65/940765/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/65/940765/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/65/940765/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/65/940765/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/65/940765/4"}}},"commit":{"parents":[{"commit":"0b6e789830a3fbe3d842d7294823f791dd75b719","subject":"pep8: Issue on Noble (U. 24.04) with pylint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0b6e789830a3fbe3d842d7294823f791dd75b719"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-02-05 08:51:05.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-02-17 14:21:18.000000000","tz":60},"subject":"Remove LinuxBridge related code","message":"Remove LinuxBridge related code\n\nWith [1] the LinuxBridge driver related code was cut from\nNeutron, let\u0027s try to do the same for networking-bagpipe.\n\n[1]:https://review.opendev.org/c/openstack/neutron/+/927216\nCloses-Bug: #2097367\n\nCloses-Bug: #2098570\nChange-Id: I8c4f857f560841a14ad86538279366bab5c9f2ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/ae61d585b651290a8786438058e5c8a1ecdb2063"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/ae61d585b651290a8786438058e5c8a1ecdb2063"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
