)]}'
{"id":"openstack%2Fnetworking-bagpipe~942584","triplet_id":"openstack%2Fnetworking-bagpipe~unmaintained%2F2023.1~I085f3343e1466f43e4318055e252347000397f0f","project":"openstack/networking-bagpipe","branch":"unmaintained/2023.1","topic":"remove-periodic-stable-from-unmaintained","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-02-28 15:28:35.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-02-28 12:31:15.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-02-28 09:47:36.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-02-28 15:28:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I085f3343e1466f43e4318055e252347000397f0f","subject":"[CI] Remove periodic stable jobs from unmaintained branch","status":"MERGED","created":"2025-02-24 13:27:53.000000000","updated":"2025-02-28 15:30:57.000000000","submitted":"2025-02-28 15:28:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942584-remove-periodic-stable-from-unmaintained","meta_rev_id":"f1865b774bce6b016ad756c11300088bc6f9018a","_number":942584,"virtual_id_number":942584,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-28 15:28:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-02-28 09:47:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-02-28 09:47:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-24 13:31:50.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-02-24 15:16:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-26 16:46:32.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-02-26 16:46:32.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"36e1a607dde4396c37abac823dc7587c119d76e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-02-24 13:27:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7cde06ee519412a406ac7729a542c4e2c1d594e1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-02-24 13:31:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ee4ec2799eb81a030ddafb60d81593eee9463426","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 15:16:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/44bf0f5f4297468c8a466b2e95102416\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93d157cbe6e64db5a76ad884676bc1df : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d111275dbad94710b653bc7da2e8e52f : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3dce276789444dfca8a63c736817dbe6 : SUCCESS in 6m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f6acaefb00f4228adee66b726939ed3 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c159f207dc64cfabbdcbda8aa32e42d : SUCCESS in 7m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d468cce623914fbfad431d479e770114 : SUCCESS in 6m 58s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/89e8e6d4684746c5992a5533ff7274ad : FAILURE in 27m 04s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/b6367034bc1b4955b79a6973f4d9fd7b : FAILURE in 8m 51s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/76b73ebb50674b1daf459ac380c7d6f3 : SUCCESS in 1h 28m 55s","accounts_in_message":[],"_revision_number":1},{"id":"2b0cac0eb21414fb1209b996418029e7ec3a170d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-02-25 14:25:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cc24c58b9a896816ca47f823dcbb091ca5163f2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 16:03:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/96ce75b7827a41f0b9816bfc4435a097\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25bcc76fd1514d23a7f49ec2be51f7a9 : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b38a7f3ebb714c07bd65d32d11deba05 : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2dfd50f37384d069184ef04fc9424b2 : SUCCESS in 6m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/114fdd411c08489c9357d076d11f1de6 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28aba11da0ba42789a9bdbca1a4fd86e : SUCCESS in 9m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d9fd732109d455684e142a991699ca4 : SUCCESS in 4m 27s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/a253ce1ebc71427597d9a212a73b210a : FAILURE in 16m 47s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/765e8db0628a4801a3b6fbadc8187f8f : FAILURE in 19m 46s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/96219f2687eb445899dc442823f842fd : SUCCESS in 1h 26m 41s","accounts_in_message":[],"_revision_number":1},{"id":"28518bf83b0faee774b872c848a9b815fe2f20a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-02-26 13:28:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0b58dfafe8d95fee141e219a2da837bc9d19f1d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 15:01:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fb2f61c93284e1e8379e883e368ab59\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0c7c7ee13b04efebc8453802fa8ffe9 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/167c4f65deb8429d8e264013c5b534c3 : SUCCESS in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/373e2a9c3f1740a393994200bed11803 : SUCCESS in 7m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a321e838efab45e188b531ef9f4d408f : SUCCESS in 8m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7cfad0ebbe443fab93febd2a6c58cb7 : SUCCESS in 11m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/994bc88d35ff4b0d9e7279d47bc4aa30 : SUCCESS in 10m 25s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/4e1649e1e333437084f22f69c95907a6 : FAILURE in 39m 38s (non-voting)\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/ba0d2ebfa8d846209ee210bddbcb37be : FAILURE in 25m 18s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/3e23f29261a749b598c0bb9ef806a07c : SUCCESS in 1h 26m 59s","accounts_in_message":[],"_revision_number":2},{"id":"ce32238e7f3531a3ab5ac256db3fbbaaa7d0581b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-02-28 09:47:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2d29c78332c55e97938ad7382e0d0c5062a45426","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 09:48:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"71a5ca71f41c23e97297fbd249665f66e6a2517e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 11:44:36.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b48c722289341b19aad08bca6aaa8e8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cbcadf831cc4294bde2d6da4af03eb1 : SUCCESS in 8m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b27518fcde246f8a2f3639a0d002ed3 : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3c0ff28e1a74cfd825bdb96ab6b6011 : SUCCESS in 6m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1e820062ba4a40f7b59b128081c386c6 : SUCCESS in 19m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0208bc454044239ab24184f37a111a6 : TIMED_OUT in 30m 37s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/d9396a1c86b64a79aba39d9b8362935c : SUCCESS in 1h 50m 29s","accounts_in_message":[],"_revision_number":2},{"id":"4ad156a088b8cd38fbdeed244d7cb3e895a19c70","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-02-28 12:31:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a5991928929f3216f7079a01824e94c2f1b50e7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 14:00:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d05b86b8f2ca4c66a3260d19b88bf712\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1119946155248538126118f2ac7016e : SUCCESS in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f31f3488a3b4f60808c52e456740c25 : SUCCESS in 8m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab37403f5cb2473eb35eb152e450d6bb : SUCCESS in 5m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/808a7f8b05a14eb58f100384295e0742 : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca23e186a9b34957acf32f26633a1be0 : SUCCESS in 10m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a24e47465a86472eaff276bba4effce0 : SUCCESS in 7m 50s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/747ef8a5b959473f96e3345f741a15b3 : FAILURE in 18m 21s (non-voting)\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/151fd9ab316e46f4b68e6298398c3ad0 : FAILURE in 17m 32s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/63b2de78a322494ea4912f0108c51f2e : SUCCESS in 1h 22m 54s","accounts_in_message":[],"_revision_number":2},{"id":"eddc5ddfbe56ed2dddc03870e63858a38ac678af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 14:01:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"aa97e97b46ebc2b245b53ba16fe1b08ee4579ae6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 15:28:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cce29ed74e440909f2a301b7bc28390\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f245e9563fe240318c8aa1a99a4e0f49 : SUCCESS in 8m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a00ab10102a34770a9349154f1aa38ae : SUCCESS in 7m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de181780a72142eaa30c67610cb30d6e : SUCCESS in 8m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d424eab58024e46951cad3226b791b5 : SUCCESS in 7m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0678b8185cc4dd6a07b2f8856e7426e : SUCCESS in 9m 23s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/0fdd7c6086b74dbf859081c5e612d27c : SUCCESS in 1h 22m 13s","accounts_in_message":[],"_revision_number":2},{"id":"5bed647b6e5ed5cb74a6d2fc483138adad9b1bfe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 15:28:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f1865b774bce6b016ad756c11300088bc6f9018a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 15:30:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/503e5fd9714c49f092b4f5884a3bdcfb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66c6f0570bb04bd39c853e271edb5b1e : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"79b2d924c9baeb4d824882c0473f0ca0060832cb","revisions":{"2c4723d853620ffde1039da34e19aa7e420b1b4e":{"kind":"REWORK","_number":1,"created":"2025-02-24 13:27:53.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/84/942584/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/84/942584/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/84/942584/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/84/942584/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/84/942584/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/84/942584/1"}}},"commit":{"parents":[{"commit":"290583923da82c1b069518f04bef10d438758de2","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/290583923da82c1b069518f04bef10d438758de2"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-02-24 13:22:17.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-02-24 13:22:17.000000000","tz":60},"subject":"[CI] Remove periodic stable jobs from unmaintained branch","message":"[CI] Remove periodic stable jobs from unmaintained branch\n\nPeriodic stable jobs are meant for stable branches and not for\nunmaintained branches. perioid-weekly is enough (if there is\nanyone, who checks its result) to keep. This patch removes all\nperiodic-stable jobs.\n\nChange-Id: I085f3343e1466f43e4318055e252347000397f0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/2c4723d853620ffde1039da34e19aa7e420b1b4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/2c4723d853620ffde1039da34e19aa7e420b1b4e"}]},"branch":"refs/heads/unmaintained/2023.1"},"79b2d924c9baeb4d824882c0473f0ca0060832cb":{"kind":"REWORK","_number":2,"created":"2025-02-26 13:28:15.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/84/942584/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/84/942584/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/84/942584/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/84/942584/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/84/942584/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/84/942584/2"}}},"commit":{"parents":[{"commit":"290583923da82c1b069518f04bef10d438758de2","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/290583923da82c1b069518f04bef10d438758de2"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-02-24 13:22:17.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-02-26 13:27:31.000000000","tz":60},"subject":"[CI] Remove periodic stable jobs from unmaintained branch","message":"[CI] Remove periodic stable jobs from unmaintained branch\n\nPeriodic stable jobs are meant for stable branches and not for\nunmaintained branches. perioid-weekly is enough (if there is\nanyone, who checks its result) to keep. This patch removes all\nperiodic-stable jobs.\n\nChange-Id: I085f3343e1466f43e4318055e252347000397f0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/79b2d924c9baeb4d824882c0473f0ca0060832cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/79b2d924c9baeb4d824882c0473f0ca0060832cb"}]},"branch":"refs/heads/unmaintained/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
