)]}'
{"id":"openstack%2Fnetworking-bagpipe~943020","triplet_id":"openstack%2Fnetworking-bagpipe~unmaintained%2Fzed~I085f3343e1466f43e4318055e252347000397f0f","project":"openstack/networking-bagpipe","branch":"unmaintained/zed","topic":"remove-periodic-stable-from-unmaintained","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-03-06 11:59:45.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-03-05 09:39:28.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}},"hashtags":[],"change_id":"I085f3343e1466f43e4318055e252347000397f0f","subject":"[CI] Remove periodic stable jobs from unmaintained branch","status":"MERGED","created":"2025-02-28 15:48:28.000000000","updated":"2025-03-06 12:01:35.000000000","submitted":"2025-03-06 11:59:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943020-remove-periodic-stable-from-unmaintained","meta_rev_id":"373623975e8ec6592d82d29bd2c27be2cf2e0348","_number":943020,"virtual_id_number":943020,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-06 11:59:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2025-03-05 09:39:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2025-03-06 10:46:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-28 17:32:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-04 19:27:37.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"aaf98b09464d7a995deff39782db85fae2a71686","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-02-28 15:48:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a42801c817a574fa04eeaa11a6ec82f62d779d5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 17:32:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a7f83de27666446cbc39b7396701e3e9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cc23015693246dfba31fc94dd09ca13 : FAILURE in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee6e90828eba45ae846a805922342f53 : FAILURE in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18efa65a92b54524abe3a9d3d2376ccb : FAILURE in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d47d5726a01c4ccfa0a3f3a9465c610a : FAILURE in 4m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31f0807b4dfe4b69a063e29f377dcfd6 : FAILURE in 6m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c1c3b11e0214e0999c6efd52c01d329 : FAILURE in 5m 11s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/20d9f43680cc49278ce7e499a88c0862 : FAILURE in 19m 08s (non-voting)\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/01f2a256bdb84bebb6150c1a251dd7c5 : FAILURE in 8m 32s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/39eed6765ae24931afc51b06733e90e3 : SUCCESS in 1h 19m 58s","accounts_in_message":[],"_revision_number":1},{"id":"4a519f40e64fd90bdbcceb9f81802b66ab0a19d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-04 16:36:24.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d59b0f61f48e244e0494be3242d2d2200d6ed664","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-04 18:09:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ef389a16c3c4d308e48087a42bddc75\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33b84bc57ba0493f9461bc5ae4a5eee8 : SUCCESS in 8m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e1c88850d9046a4ae648330f78e1fb4 : SUCCESS in 7m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77a93fe071524971bd34b4ce2c62089b : SUCCESS in 6m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/804e02efe3c1433a86f6b92678100552 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db6671b285fa49ac8ca10d2051df543a : SUCCESS in 9m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99ce424dfee7494f94a1de7b85350cb9 : SUCCESS in 7m 19s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/a51fbf4d5648413fbbbc8c3ac65c3482 : FAILURE in 25m 48s (non-voting)\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/3fc7390d22da4beca996a557feef5480 : FAILURE in 9m 15s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/9db88e33860841b98f229fba4d19e45b : SUCCESS in 1h 16m 03s","accounts_in_message":[],"_revision_number":2},{"id":"87f5a0dbca96c64fd4a8913cbe3940b7eeb3c968","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-03-05 09:39:28.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"be28995426148ad74db7a5fdb28286afdc26817f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-03-06 10:46:18.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3165c95ce937eb2d8607ebcf427c8d315a1ea597","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 10:47:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fdc8c6a78771dce2dfc57965f271112a5a9ec47a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 11:59:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7f28266e3a24524a9754c1a588fd2f8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4208a7b06ec34991bbd933e97da26227 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd8a0c0f361e487fbf0253430a37dad0 : SUCCESS in 6m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da3671131fac44e6af3839cf3affff86 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/678bedcc462d4229af085b4829336c94 : SUCCESS in 10m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/615a1c9525b344538cfc183a8447f420 : SUCCESS in 6m 42s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/2af9df26b7694245ae6a845832ed9470 : SUCCESS in 1h 10m 57s","accounts_in_message":[],"_revision_number":2},{"id":"861de0a1f88b51e82e2c1a9799f501bb83fd948b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 11:59:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"373623975e8ec6592d82d29bd2c27be2cf2e0348","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 12:01:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15dc9484197b4eb1823e169c7d6e5265\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a536c301ac0f495e9fe26970ec18b673 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4570f681ec7873d854dbca32508008c292315491","revisions":{"6ad598660e9fd438bed8ebd2f269e99d6791ac60":{"kind":"REWORK","_number":1,"created":"2025-02-28 15:48:28.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/20/943020/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/20/943020/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/20/943020/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/20/943020/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/20/943020/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/20/943020/1"}}},"commit":{"parents":[{"commit":"6c84078343e791454f7f6f622138cc19aec7e7be","subject":"Update .gitreview for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/6c84078343e791454f7f6f622138cc19aec7e7be"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-02-24 13:22:17.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-02-28 15:47:53.000000000","tz":60},"subject":"[CI] Remove periodic stable jobs from unmaintained branch","message":"[CI] Remove periodic stable jobs from unmaintained branch\n\nPeriodic stable jobs are meant for stable branches and not for\nunmaintained branches. perioid-weekly is enough (if there is\nanyone, who checks its result) to keep. This patch removes all\nperiodic-stable jobs.\n\nConflicts:\n  .zuul.yaml\n\nChange-Id: I085f3343e1466f43e4318055e252347000397f0f\n(cherry picked from commit 79b2d924c9baeb4d824882c0473f0ca0060832cb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/6ad598660e9fd438bed8ebd2f269e99d6791ac60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/6ad598660e9fd438bed8ebd2f269e99d6791ac60"}]},"branch":"refs/heads/unmaintained/zed"},"4570f681ec7873d854dbca32508008c292315491":{"kind":"REWORK","_number":2,"created":"2025-03-04 16:36:24.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/20/943020/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/20/943020/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/20/943020/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/20/943020/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/20/943020/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/20/943020/2"}}},"commit":{"parents":[{"commit":"6c84078343e791454f7f6f622138cc19aec7e7be","subject":"Update .gitreview for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/6c84078343e791454f7f6f622138cc19aec7e7be"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-02-24 13:22:17.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-03-04 16:34:59.000000000","tz":60},"subject":"[CI] Remove periodic stable jobs from unmaintained branch","message":"[CI] Remove periodic stable jobs from unmaintained branch\n\nPeriodic stable jobs are meant for stable branches and not for\nunmaintained branches. perioid-weekly is enough (if there is\nanyone, who checks its result) to keep. This patch removes all\nperiodic-stable jobs.\n\nThis patch also removes networking-sfc from required projects in\nzuul.yaml as unmaintained/zed and older branches do not exist\nanymore for this repository.\n\nConflicts:\n  .zuul.yaml\n\nChange-Id: I085f3343e1466f43e4318055e252347000397f0f\n(cherry picked from commit 79b2d924c9baeb4d824882c0473f0ca0060832cb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/4570f681ec7873d854dbca32508008c292315491"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/4570f681ec7873d854dbca32508008c292315491"}]},"branch":"refs/heads/unmaintained/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
