)]}'
{"id":"openstack%2Fnetworking-bagpipe~943115","triplet_id":"openstack%2Fnetworking-bagpipe~master~Ie7dc0aaaab3b7ad03f5546dc876520b99a15a296","project":"openstack/networking-bagpipe","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-03-06 16:58:48.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-03-06 15:25:47.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"Ie7dc0aaaab3b7ad03f5546dc876520b99a15a296","subject":"Drop removed tables for linux bridge plugin","status":"MERGED","created":"2025-03-02 05:00:12.000000000","updated":"2025-03-06 16:59:52.000000000","submitted":"2025-03-06 16:58:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943115","meta_rev_id":"fc5caec0484abe7ab846e12e7813d6f960666a17","_number":943115,"virtual_id_number":943115,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-06 16:58:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2025-03-03 12:30:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-03-06 15:25:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-02 15:38:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2025-03-06 15:25:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-02 07:55:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-02 15:38:30.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-03-03 12:30:05.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-03-05 10:07:26.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"7c6b50a88f21be9711efa0ca696ac61989cd7f40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-02 05:00:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"97120061b345ceb36a30aa30d2a4a2ccda580d74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-02 05:10:50.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4c7eb7d3fce343f6db92182907118b675d37cd7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-02 05:11:04.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"e51dcb06eaec31988ec2eb158620d3bc7da65085","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-02 06:28:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a2953d2a799205b92396f8028b9b4b96a72fce22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-02 07:55:20.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2832a0005927426191e0bdd530df7b97\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47f380a9448147698266b350c4702357 : FAILURE in 6m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7810d44046004a7da45b61f8291185a1 : SUCCESS in 6m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/58c60364c2924bdf8e3767bd51f7492a : SUCCESS in 9m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62e720c7ba584562923fbbad5b32510e : SUCCESS in 10m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b2c4bc5f6574542ae1f31ed6f8b9208 : SUCCESS in 11m 28s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/5247a319aace404899ff195087369d83 : FAILURE in 20m 48s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/61c49cc0a56f42f6a437f91abdda25e4 : FAILURE in 19m 41s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/4f6983619a744d0ea83c6ac6dcfbeaef : SUCCESS in 1h 26m 45s","accounts_in_message":[],"_revision_number":4},{"id":"a96a3c3aca59f88d1515a86173e1a5de5691fe3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-02 13:48:14.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"b2fd7f205eef6ad07376a9c55756f0171e8aae54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-02 13:49:33.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"7a9f1196b36d64e46c680d5c15651f8d8c0d2bb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-02 14:59:02.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fdd34e7947a54bfe9e9a58aad44b3687\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38c3196235954aebbd6a13e941d4ac95 : SUCCESS in 8m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0f0ba8d3314457b979a94c3003a0cd6 : SUCCESS in 9m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/532c106f4b214165a4ef2093a52e052c : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1d9b779b574491ca16488c346fd0a6e : SUCCESS in 12m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eaf3e96f992c4395beb7516f75e0a00a : SUCCESS in 7m 33s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/bdc7ee1f7bdd4688947f50d81b32e167 : FAILURE in 27m 29s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/09ee6e1ee5a44d1c96d6ae66f9839489 : FAILURE in 12m 35s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/f87b47daf81b4bf697910055a97a5ddb : FAILURE in 1h 08m 36s","accounts_in_message":[],"_revision_number":6},{"id":"5d726ff37ac34607e650b3cebf8faf4bcc813c70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-02 15:16:00.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"7bbbd1e01b39ea81191ab53ae39e73defe2185ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-02 15:18:03.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"5136e809a5846d07aec3280447a3050de73522ce","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-02 15:38:30.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"3d39d5b8f408ef63009255c895825abbb6f359d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-02 16:53:54.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f74b1b1ac8f149bf8c6118270fcbf659\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b881f5fafa8437e94c78cdf5c5c51f8 : SUCCESS in 9m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3bf55c4ca9b54ca58e4e0a5d4f5f65c7 : SUCCESS in 6m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ddd2b41a96fa4c7281ad2724e02dd17c : SUCCESS in 9m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9e1f9c0efe14845914defa58439fb4c : SUCCESS in 9m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f77e2b4b2f1416e8b2987d59d386918 : SUCCESS in 9m 57s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/7740d4ae916942debbee80e3480253dc : SUCCESS in 1h 20m 38s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/583a8929e81a4e4d8f93be334e256e3e : FAILURE in 7m 05s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/cf004649d18a485fa1d9804aeb05ab55 : SUCCESS in 1h 28m 24s","accounts_in_message":[],"_revision_number":8},{"id":"3914de462d4817072475ea0b8584252b64dda7e5","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-03-03 12:30:05.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"66d70c0ab39957fe0c1deae5e1fe365d0ce4f947","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-03-06 15:25:47.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"0032b74e682533b92fd6b628bfc6cb4a4e99cd65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 15:26:18.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"4857e18a7315a7641cea4af8cb71e4c33832e685","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 16:58:48.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3360d7ef1314d30a2313b235ef7a12b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/106e6633ec184d5cb454d2c8855b12b9 : SUCCESS in 8m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23b69a33a0a044299c2c6027c2e3d6c8 : SUCCESS in 7m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cc94cadd52e74c99b229291f1a7c3cf2 : SUCCESS in 9m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/057cad91a25b42a6a027d95dc72a7490 : SUCCESS in 9m 17s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/5d143c5e667940669f1b67020889e56c : SUCCESS in 1h 32m 00s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/7660c9d3642646ef82b6bf7b76d4e63e : SUCCESS in 1h 30m 41s","accounts_in_message":[],"_revision_number":8},{"id":"d2b9850d613a822a2ab0898ecf68c11a6632960b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 16:58:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"fc5caec0484abe7ab846e12e7813d6f960666a17","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 16:59:52.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90bfcff5ae254ac09d41ca8f93870e0f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a44bd043b22e4d32abe3bd39fefacf3e : SUCCESS in 46s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"c48cdb2a5f0a41a7bfebfd65254b7b5c4ab75563","revisions":{"04d0a3ab1620cea6c0407863d62d3397dac7a55b":{"kind":"REWORK","_number":1,"created":"2025-03-02 05:00:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/943115/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/15/943115/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/1"}}},"commit":{"parents":[{"commit":"db7fb00fc4bdb5ae0875735bcee9b0434af5b745","subject":"Remove unused Babel","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/db7fb00fc4bdb5ae0875735bcee9b0434af5b745"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-02 04:59:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-02 04:59:35.000000000","tz":540},"subject":"Drop removed tables for linux bridge plugin","message":"Drop removed tables for linux bridge plugin\n\nThis is follow-up of ae61d585b651290a8786438058e5c8a1ecdb2063 and\nensures that unnecessary tables are dropped during upgrade.\n\nChange-Id: Ie7dc0aaaab3b7ad03f5546dc876520b99a15a296\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/04d0a3ab1620cea6c0407863d62d3397dac7a55b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/04d0a3ab1620cea6c0407863d62d3397dac7a55b"}]},"branch":"refs/heads/master"},"3b868a2d9ef82cc7b1e289772dc93d704edc6272":{"kind":"REWORK","_number":2,"created":"2025-03-02 05:10:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/943115/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/15/943115/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/2"}}},"commit":{"parents":[{"commit":"db7fb00fc4bdb5ae0875735bcee9b0434af5b745","subject":"Remove unused Babel","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/db7fb00fc4bdb5ae0875735bcee9b0434af5b745"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-02 04:59:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-02 05:10:50.000000000","tz":0},"subject":"Drop removed tables for linux bridge plugin","message":"Drop removed tables for linux bridge plugin\n\nThis is follow-up of ae61d585b651290a8786438058e5c8a1ecdb2063 and\nensures that unnecessary tables are dropped during upgrade.\n\nChange-Id: Ie7dc0aaaab3b7ad03f5546dc876520b99a15a296\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/3b868a2d9ef82cc7b1e289772dc93d704edc6272"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/3b868a2d9ef82cc7b1e289772dc93d704edc6272"}]},"branch":"refs/heads/master"},"c100b3e946d558e9463e88c08f728400b0c2cf57":{"kind":"REWORK","_number":3,"created":"2025-03-02 05:11:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/943115/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/15/943115/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/3"}}},"commit":{"parents":[{"commit":"db7fb00fc4bdb5ae0875735bcee9b0434af5b745","subject":"Remove unused Babel","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/db7fb00fc4bdb5ae0875735bcee9b0434af5b745"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-02 04:59:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-02 05:11:04.000000000","tz":0},"subject":"Drop removed tables for linux bridge plugin","message":"Drop removed tables for linux bridge plugin\n\nThis is follow-up of ae61d585b651290a8786438058e5c8a1ecdb2063 and\nensures that unnecessary tables are dropped during upgrade.\n\nChange-Id: Ie7dc0aaaab3b7ad03f5546dc876520b99a15a296\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c100b3e946d558e9463e88c08f728400b0c2cf57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c100b3e946d558e9463e88c08f728400b0c2cf57"}]},"branch":"refs/heads/master"},"0d6add7c49b9581be4a566cc359e2d2fc0728528":{"kind":"REWORK","_number":4,"created":"2025-03-02 06:28:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/943115/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/15/943115/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/4"}}},"commit":{"parents":[{"commit":"db7fb00fc4bdb5ae0875735bcee9b0434af5b745","subject":"Remove unused Babel","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/db7fb00fc4bdb5ae0875735bcee9b0434af5b745"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-02 04:59:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-02 06:27:53.000000000","tz":540},"subject":"Drop removed tables for linux bridge plugin","message":"Drop removed tables for linux bridge plugin\n\nThis is follow-up of ae61d585b651290a8786438058e5c8a1ecdb2063 and\nensures that unnecessary tables are dropped during upgrade.\n\nChange-Id: Ie7dc0aaaab3b7ad03f5546dc876520b99a15a296\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0d6add7c49b9581be4a566cc359e2d2fc0728528"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/0d6add7c49b9581be4a566cc359e2d2fc0728528"}]},"branch":"refs/heads/master"},"72c6c848c0709e37a03bf8738639c86d20d0d9f7":{"kind":"REWORK","_number":5,"created":"2025-03-02 13:48:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/943115/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/15/943115/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/5"}}},"commit":{"parents":[{"commit":"db7fb00fc4bdb5ae0875735bcee9b0434af5b745","subject":"Remove unused Babel","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/db7fb00fc4bdb5ae0875735bcee9b0434af5b745"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-02 04:59:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-02 13:48:14.000000000","tz":0},"subject":"Drop removed tables for linux bridge plugin","message":"Drop removed tables for linux bridge plugin\n\nThis is follow-up of ae61d585b651290a8786438058e5c8a1ecdb2063 and\nensures that unnecessary tables are dropped during upgrade.\n\nChange-Id: Ie7dc0aaaab3b7ad03f5546dc876520b99a15a296\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/72c6c848c0709e37a03bf8738639c86d20d0d9f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/72c6c848c0709e37a03bf8738639c86d20d0d9f7"}]},"branch":"refs/heads/master"},"51e2f0a2746d49a8aff9df9bd182fa0b23e03fd8":{"kind":"REWORK","_number":6,"created":"2025-03-02 13:49:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/943115/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/15/943115/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/6"}}},"commit":{"parents":[{"commit":"db7fb00fc4bdb5ae0875735bcee9b0434af5b745","subject":"Remove unused Babel","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/db7fb00fc4bdb5ae0875735bcee9b0434af5b745"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-02 04:59:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-02 13:49:32.000000000","tz":0},"subject":"Drop removed tables for linux bridge plugin","message":"Drop removed tables for linux bridge plugin\n\nThis is follow-up of ae61d585b651290a8786438058e5c8a1ecdb2063 and\nensures that unnecessary tables are dropped during upgrade.\n\nChange-Id: Ie7dc0aaaab3b7ad03f5546dc876520b99a15a296\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/51e2f0a2746d49a8aff9df9bd182fa0b23e03fd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/51e2f0a2746d49a8aff9df9bd182fa0b23e03fd8"}]},"branch":"refs/heads/master"},"b9675f027b8702c141fe55e98e973d539a3ce7c1":{"kind":"REWORK","_number":7,"created":"2025-03-02 15:16:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/943115/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/15/943115/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/7"}}},"commit":{"parents":[{"commit":"db7fb00fc4bdb5ae0875735bcee9b0434af5b745","subject":"Remove unused Babel","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/db7fb00fc4bdb5ae0875735bcee9b0434af5b745"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-02 04:59:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-02 15:15:52.000000000","tz":540},"subject":"Drop removed tables for linux bridge plugin","message":"Drop removed tables for linux bridge plugin\n\nThis is follow-up of ae61d585b651290a8786438058e5c8a1ecdb2063 and\nensures that unnecessary tables are dropped during upgrade.\n\nChange-Id: Ie7dc0aaaab3b7ad03f5546dc876520b99a15a296\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b9675f027b8702c141fe55e98e973d539a3ce7c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b9675f027b8702c141fe55e98e973d539a3ce7c1"}]},"branch":"refs/heads/master"},"c48cdb2a5f0a41a7bfebfd65254b7b5c4ab75563":{"kind":"REWORK","_number":8,"created":"2025-03-02 15:18:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/943115/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/15/943115/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/15/943115/8"}}},"commit":{"parents":[{"commit":"db7fb00fc4bdb5ae0875735bcee9b0434af5b745","subject":"Remove unused Babel","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/db7fb00fc4bdb5ae0875735bcee9b0434af5b745"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-02 04:59:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-02 15:17:55.000000000","tz":540},"subject":"Drop removed tables for linux bridge plugin","message":"Drop removed tables for linux bridge plugin\n\nThis is follow-up of ae61d585b651290a8786438058e5c8a1ecdb2063 and\nensures that unnecessary tables are dropped during upgrade.\n\nChange-Id: Ie7dc0aaaab3b7ad03f5546dc876520b99a15a296\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c48cdb2a5f0a41a7bfebfd65254b7b5c4ab75563"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c48cdb2a5f0a41a7bfebfd65254b7b5c4ab75563"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
