)]}'
{"id":"openstack%2Fnetworking-bagpipe~943126","triplet_id":"openstack%2Fnetworking-bagpipe~master~I9485fa055aed18bef1128101ca12b179c968bb23","project":"openstack/networking-bagpipe","branch":"master","topic":"remove-linuxbridge-agent","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-03-05 16:37:35.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-03-05 16:37:35.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-03-05 14:44:39.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I9485fa055aed18bef1128101ca12b179c968bb23","subject":"Drop remaining code related to linux bridge driver","status":"MERGED","created":"2025-03-03 01:42:36.000000000","updated":"2025-03-05 16:38:41.000000000","submitted":"2025-03-05 16:37:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943126-remove-linuxbridge-agent","meta_rev_id":"b430fafc46bf36be8c5ce34d2c68aa4146fa9e8c","_number":943126,"virtual_id_number":943126,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-05 16:37:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2025-03-05 10:07:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2025-03-05 14:44:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-03 08:06:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2025-03-05 14:44:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-03 05:17:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-03 08:06:51.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-03-05 10:07:05.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-03-05 10:07:39.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-03-05 10:10:51.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"eadd743fc1bdaa65fc3b6443fb88715f8edb4ee1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-03 01:42:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb9c48c5cbbec5f0690aafddf6a7e0688945ff6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-03 02:48:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dabd60e04b4490ddd46c2b16fa36d0cfd954afbe","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-03 02:51:32.000000000","message":"Topic set to remove_linux_bridge","accounts_in_message":[],"_revision_number":2},{"id":"9762a8764bc2d598351973f0c8b818bfd6c26187","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-03 02:51:57.000000000","message":"Topic remove_linux_bridge removed","accounts_in_message":[],"_revision_number":2},{"id":"44f39f347c6aeeeb35984edb43d5814817b56dcc","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-03 02:52:03.000000000","message":"Topic set to remove-linuxbridge-agent","accounts_in_message":[],"_revision_number":2},{"id":"47cf50418081d75d644190cfad67352a0ef13551","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-03 03:56:00.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"112250033bd4dbff09b767afb4c19d3d31053c80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 05:17:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db69a481e77a437db3f0ed1ef057be46\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/192023ad352b4047893344566512c614 : SUCCESS in 6m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18f3ab5d355a401a8d27386f285450ab : SUCCESS in 6m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7903fee84be4429dbfe6d792c7293d8a : SUCCESS in 9m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2861b5b753084419943cc3c5a017e42c : SUCCESS in 8m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7cf494c1d78941c8b454a3939411d518 : SUCCESS in 10m 49s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/0e2f2a08877c437c82a5a932512a8424 : SUCCESS in 1h 20m 35s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/ef2bb5e3349e45efac49cd755bcdc46c : FAILURE in 13m 47s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/3a23d7730de742e498dab611fd48bce3 : SUCCESS in 1h 17m 47s","accounts_in_message":[],"_revision_number":3},{"id":"899a80274cef3ab94fd8280a999a8b7b37052220","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-03 08:06:51.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"6152b3a6a92cd4ca8abeddf159ab9c4912ff56d1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-03-05 10:07:05.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"39c2ad05c938a28c6f026ba0e60dd51f80b2a57c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-03-05 14:44:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"50fd0411182fc71c9952332d8e2ce517e60f9ca0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 14:45:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"88b6a67f4439a79fa87b1bc6942f4eb589243ad9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 16:37:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48510213c99d4075a07add0e6833dba0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/154a7cc9b8994f799716fd6e0f2cc928 : SUCCESS in 8m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6abe99d6de69403c930c8269519d39d5 : SUCCESS in 7m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5f83fc467f3d4dd5abdb418c9c4b2a3b : SUCCESS in 8m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f56761f19704c2581c6d8aef301e013 : SUCCESS in 8m 16s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/d813c7ea9ba14910bb1359f5f7d9546d : SUCCESS in 1h 47m 31s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/01e7534ee08b4c1cbfd650f5675c06be : SUCCESS in 1h 32m 00s","accounts_in_message":[],"_revision_number":3},{"id":"66192421331eb9fe48913eac58c6c1ae0d3bb919","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 16:37:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b430fafc46bf36be8c5ce34d2c68aa4146fa9e8c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 16:38:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2420403a82e45fd8424f5fcac797d2c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67442e505df7400fb08b7013f1f82d1a : SUCCESS in 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b16f0558e2c822ddf3548d706f989b20786ca96b","revisions":{"bcf0ad2aabf00c6aa2d1da568622e7ff20f78984":{"kind":"REWORK","_number":1,"created":"2025-03-03 01:42:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/943126/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/26/943126/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/943126/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/943126/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/943126/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/943126/1"}}},"commit":{"parents":[{"commit":"d21d3a22e720386390c17a5745bcab8c33fa2a77","subject":"Uncap psutil","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/d21d3a22e720386390c17a5745bcab8c33fa2a77"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-03 01:41:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-03 01:42:07.000000000","tz":540},"subject":"Drop remaining code related to linuxb bridge driver","message":"Drop remaining code related to linuxb bridge driver\n\nThese were overlooked during the previous attempt[1].\n\n[1] ae61d585b651290a8786438058e5c8a1ecdb2063\n\nChange-Id: I9485fa055aed18bef1128101ca12b179c968bb23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/bcf0ad2aabf00c6aa2d1da568622e7ff20f78984"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/bcf0ad2aabf00c6aa2d1da568622e7ff20f78984"}]},"branch":"refs/heads/master"},"afab4afa043c1eba8780ca36d9c3ca524d32534e":{"kind":"REWORK","_number":2,"created":"2025-03-03 02:48:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/943126/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/26/943126/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/943126/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/943126/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/943126/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/943126/2"}}},"commit":{"parents":[{"commit":"d21d3a22e720386390c17a5745bcab8c33fa2a77","subject":"Uncap psutil","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/d21d3a22e720386390c17a5745bcab8c33fa2a77"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-03 01:41:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-03 02:48:32.000000000","tz":540},"subject":"Drop remaining code related to linuxb bridge driver","message":"Drop remaining code related to linuxb bridge driver\n\nThese were overlooked during the previous attempt[1].\n\n[1] ae61d585b651290a8786438058e5c8a1ecdb2063\n\nChange-Id: I9485fa055aed18bef1128101ca12b179c968bb23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/afab4afa043c1eba8780ca36d9c3ca524d32534e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/afab4afa043c1eba8780ca36d9c3ca524d32534e"}]},"branch":"refs/heads/master"},"b16f0558e2c822ddf3548d706f989b20786ca96b":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-03-03 03:56:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/943126/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/26/943126/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/943126/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/943126/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/943126/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/943126/3"}}},"commit":{"parents":[{"commit":"d21d3a22e720386390c17a5745bcab8c33fa2a77","subject":"Uncap psutil","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/d21d3a22e720386390c17a5745bcab8c33fa2a77"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-03 01:41:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-03 03:56:00.000000000","tz":0},"subject":"Drop remaining code related to linux bridge driver","message":"Drop remaining code related to linux bridge driver\n\nThese were overlooked during the previous attempt[1].\n\n[1] ae61d585b651290a8786438058e5c8a1ecdb2063\n\nChange-Id: I9485fa055aed18bef1128101ca12b179c968bb23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b16f0558e2c822ddf3548d706f989b20786ca96b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b16f0558e2c822ddf3548d706f989b20786ca96b"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
