)]}'
{"id":"openstack%2Fnetworking-bagpipe~943367","triplet_id":"openstack%2Fnetworking-bagpipe~unmaintained%2Fyoga~I085f3343e1466f43e4318055e252347000397f0f","project":"openstack/networking-bagpipe","branch":"unmaintained/yoga","topic":"remove-periodic-stable-from-unmaintained","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-03-06 10:50:38.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2025-03-06 10:50:38.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-03-06 09:36:07.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}},"hashtags":[],"change_id":"I085f3343e1466f43e4318055e252347000397f0f","subject":"[CI] Remove periodic stable jobs from unmaintained branch","status":"MERGED","created":"2025-03-05 09:28:41.000000000","updated":"2025-03-06 10:52:31.000000000","submitted":"2025-03-06 10:50:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943367-remove-periodic-stable-from-unmaintained","meta_rev_id":"0b9c5db9b0ef1fb866762942e9a347e06ef8850e","_number":943367,"virtual_id_number":943367,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-06 10:50:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2025-03-06 09:36:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2025-03-06 09:38:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-05 09:39:55.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-03-05 11:08:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-05 16:12:45.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"c79c7f1e21dc21e71982e78d5251cbce76007588","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-05 09:28:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91572af44b59f616c7dd57955d5624b6ae0ea9a0","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-03-05 09:39:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"080ee093ee5988f8abab0d55459499922b73c72f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 11:08:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d0294270c3684c4ba745a9597339cb91\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c783904a48d485ba4406f67a9334ff2 : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f685f3fda60049feb909aa99f0895f38 : SUCCESS in 8m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71c9cf20dc3e4890a7f78717513571fa : FAILURE in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f41e6386730d4ca1b98dae79814d7479 : SUCCESS in 9m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d3a0e046753491fb337b13f27b81125 : FAILURE in 3m 41s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4077c0249cc42379a3faef0ed6ace0f : FAILURE in 5m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2734a6440cbd4d92ae11769ffd7ffb10 : FAILURE in 5m 27s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/ca6b2e9058f043c1925c20c1cbadc959 : FAILURE in 43m 23s (non-voting)\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/61192cb84f3d4c6bb0448a611cc56b00 : FAILURE in 18m 42s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/f28b520e1a1d479089caeea75a1fd704 : SUCCESS in 1h 32m 42s","accounts_in_message":[],"_revision_number":1},{"id":"c23d697e7b2c9cc2f3a4b6a4fbcadc34d4b8d664","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-05 12:08:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"662f0b0c66dd1e673f423c2d64979ac87e0c5b45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 13:33:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0905b42dcda24911aefb00f188946d19\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c14d4f3a788d4f4fbf3711a4e792025f : SUCCESS in 10m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61e5ffce074f47fb8a3a2aa952c16e6d : SUCCESS in 11m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c45834f7af8444cb66586ecfc8963df : FAILURE in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/324ce41adab94c8a9aecacf9442cb9db : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e01db8e42995452ab6a16c70ef39860d : SUCCESS in 3m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f1582a7cde846d6954d5c8db7243fa8 : FAILURE in 3m 04s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8bfea4a5cb7d459286989a93e0921e8f : FAILURE in 4m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ab104ad505545d19e5e641f8eea0ab3 : FAILURE in 7m 40s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/acf43567c1b84473aeeb5e28dcc4d798 : FAILURE in 41m 49s (non-voting)\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/7943945a1b09430a9328a8f0b7a2adf2 : FAILURE in 6m 48s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/25e309b9389e45e9a5da2afc9fd706fb : SUCCESS in 1h 19m 10s","accounts_in_message":[],"_revision_number":2},{"id":"f9ccb9af4ef652dc7efadead6f9bc6182b2f4b86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-05 15:11:07.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"1ab25aa255bc794d7bd2ee0573b98307bf699fe7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-05 15:59:33.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8af11b636157d09d323f025216b07a709e5051c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 17:16:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53c1e4a6cb174c779876ca02598b0932\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4395460bf7474a35b597fdb10b3c7e56 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e37b0c0ea4494653ae4246498a507576 : SUCCESS in 7m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60210fa366fe406b9819f68081586298 : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e986f5590eb74f1aaddeb41fe31ac718 : SUCCESS in 8m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25a4e8d62a0c4cc5b1bf9d04da604129 : SUCCESS in 3m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/48ca4a0c4dfe4f56b239c98eb645da5b : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/828f254ac7194ebb804a0e063259e834 : SUCCESS in 6m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/390deab1631844b5bcccef8d038175ce : SUCCESS in 7m 13s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/2d1c046fc72643cd804e2d07e855d482 : FAILURE in 42m 34s (non-voting)\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/c1b59ca86aec4f1e934e990e66c56809 : FAILURE in 9m 17s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/2006f0fc61154c3ab8916e38e5ca74ba : SUCCESS in 45m 16s","accounts_in_message":[],"_revision_number":4},{"id":"f093dce7ccfaaddd34584d39b16f07dab29902c4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-03-06 09:36:07.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"53d65ec9971f0493f72ea2976bdc1e0cc7b6e8e1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-03-06 09:38:49.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"cc1dd0d47e6e3a307fc896878454b06a79929ef3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 09:39:24.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2239533e8f7d04ba2768a6aaec2fb810f84023d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 10:50:38.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4468942c84144f74bb1ce807c0d9cc8d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbd5a3433bb24d979b8bee0e8f3abfa5 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddb3376f9f254e899ac97b7d4810eaf7 : SUCCESS in 8m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c13c2b9c67eb47d1b3691c2810c4f034 : SUCCESS in 7m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5877ef0243aa4c688e9bdaefd5a0640b : SUCCESS in 8m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4981b92703d472888258a07836d0deb : SUCCESS in 3m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/73271f6c30ae4439abae21e642e84337 : SUCCESS in 7m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05cb1eac3e964c4b86b6c5b454b24128 : SUCCESS in 6m 12s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/6c28b7ebd5034c9091354a8beebe6343 : SUCCESS in 1h 10m 46s","accounts_in_message":[],"_revision_number":4},{"id":"e2a0afef7ddb14e46ffab4c455bcb4bf1c1ce6c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 10:50:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"0b9c5db9b0ef1fb866762942e9a347e06ef8850e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 10:52:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3b1a9266516449eb6fa81df4c4e522c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d496c557fe0b4194a7da1bb9ccf17846 : SUCCESS in 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b88f693edf3548e38ea9d58aa6ed82e9 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e93b33d044ce91eb1afb9f6c491df8cd0bae3844","revisions":{"7cc2f88e50d82595ff0f4d7e33246b64f3b8d57d":{"kind":"REWORK","_number":1,"created":"2025-03-05 09:28:41.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/67/943367/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/67/943367/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/67/943367/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/67/943367/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/67/943367/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/67/943367/1"}}},"commit":{"parents":[{"commit":"c8f6a7de016b957b7b7c57b996507f1054a4b1dc","subject":"Update .gitreview for unmaintained/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c8f6a7de016b957b7b7c57b996507f1054a4b1dc"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-02-24 13:22:17.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-03-05 09:21:51.000000000","tz":60},"subject":"[CI] Remove periodic stable jobs from unmaintained branch","message":"[CI] Remove periodic stable jobs from unmaintained branch\n\nPeriodic stable jobs are meant for stable branches and not for\nunmaintained branches. perioid-weekly is enough (if there is\nanyone, who checks its result) to keep. This patch removes all\nperiodic-stable jobs.\n\nThis patch also removes networking-sfc from required projects in\nzuul.yaml as unmaintained/zed and older branches do not exist\nanymore for this repository.\n\nConflicts:\n  .zuul.yaml\n\nChange-Id: I085f3343e1466f43e4318055e252347000397f0f\n(cherry picked from commit 79b2d924c9baeb4d824882c0473f0ca0060832cb)\n(cherry picked from commit 4570f681ec7873d854dbca32508008c292315491)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7cc2f88e50d82595ff0f4d7e33246b64f3b8d57d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/7cc2f88e50d82595ff0f4d7e33246b64f3b8d57d"}]},"branch":"refs/heads/unmaintained/yoga"},"f57ca887eade5b93dc2bfe9bb4fa836fe1c1c2f5":{"kind":"REWORK","_number":2,"created":"2025-03-05 12:08:41.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/67/943367/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/67/943367/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/67/943367/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/67/943367/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/67/943367/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/67/943367/2"}}},"commit":{"parents":[{"commit":"c8f6a7de016b957b7b7c57b996507f1054a4b1dc","subject":"Update .gitreview for unmaintained/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c8f6a7de016b957b7b7c57b996507f1054a4b1dc"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-02-24 13:22:17.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-03-05 12:06:58.000000000","tz":60},"subject":"[CI] Remove periodic stable jobs from unmaintained branch","message":"[CI] Remove periodic stable jobs from unmaintained branch\n\nPeriodic stable jobs are meant for stable branches and not for\nunmaintained branches. perioid-weekly is enough (if there is\nanyone, who checks its result) to keep. This patch removes all\nperiodic-stable jobs.\n\nThis patch also removes networking-sfc from required projects in\nzuul.yaml as unmaintained/zed and older branches do not exist\nanymore for this repository.\n\nAlso include this patch to unblock the gate: Cap setuptools \u003c71.0.0\n\npy39 jobs (on ubuntu-focal) started to fail (see the relevant github\nissue [1]) due to recent virtualenv release (20.26.4; which bundles\nsetuptools).\n\nsetuptools is bundled in virtualenv, so it has to be capped via the\nvirtualenv package. tox also needed to be capped (\u003c4) as gate uses\ntox 3.28.0, but with capping virtualenv we pull in latest tox as well,\nwhich would cause other errors.\n\nConflicts:\n  .zuul.yaml\n\n[1] https://github.com/pypa/setuptools/issues/4483\n\nChange-Id: I085f3343e1466f43e4318055e252347000397f0f\n(cherry picked from commit 79b2d924c9baeb4d824882c0473f0ca0060832cb)\n(cherry picked from commit 4570f681ec7873d854dbca32508008c292315491)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f57ca887eade5b93dc2bfe9bb4fa836fe1c1c2f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/f57ca887eade5b93dc2bfe9bb4fa836fe1c1c2f5"}]},"branch":"refs/heads/unmaintained/yoga"},"d228c29f4c273bb5e730535e1d0fc925f0cd6230":{"kind":"REWORK","_number":3,"created":"2025-03-05 15:11:07.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/67/943367/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/67/943367/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/67/943367/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/67/943367/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/67/943367/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/67/943367/3"}}},"commit":{"parents":[{"commit":"c8f6a7de016b957b7b7c57b996507f1054a4b1dc","subject":"Update .gitreview for unmaintained/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c8f6a7de016b957b7b7c57b996507f1054a4b1dc"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-02-24 13:22:17.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-03-05 15:09:34.000000000","tz":60},"subject":"[CI] Remove periodic stable jobs from unmaintained branch","message":"[CI] Remove periodic stable jobs from unmaintained branch\n\nPeriodic stable jobs are meant for stable branches and not for\nunmaintained branches. perioid-weekly is enough (if there is\nanyone, who checks its result) to keep. This patch removes all\nperiodic-stable jobs.\n\nThis patch also removes networking-sfc from required projects in\nzuul.yaml as unmaintained/zed and older branches do not exist\nanymore for this repository. Furthermore, in this branch we have\nto cap as well networking-sfc to avoid incompatibilities with\ncurrent neutron version.\n\nConflicts:\n  .zuul.yaml\n\nChange-Id: I085f3343e1466f43e4318055e252347000397f0f\n(cherry picked from commit 79b2d924c9baeb4d824882c0473f0ca0060832cb)\n(cherry picked from commit 4570f681ec7873d854dbca32508008c292315491)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/d228c29f4c273bb5e730535e1d0fc925f0cd6230"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/d228c29f4c273bb5e730535e1d0fc925f0cd6230"}]},"branch":"refs/heads/unmaintained/yoga"},"e93b33d044ce91eb1afb9f6c491df8cd0bae3844":{"kind":"REWORK","_number":4,"created":"2025-03-05 15:59:33.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/67/943367/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/67/943367/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/67/943367/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/67/943367/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/67/943367/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/67/943367/4"}}},"commit":{"parents":[{"commit":"c8f6a7de016b957b7b7c57b996507f1054a4b1dc","subject":"Update .gitreview for unmaintained/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c8f6a7de016b957b7b7c57b996507f1054a4b1dc"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-02-24 13:22:17.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-03-05 15:58:26.000000000","tz":60},"subject":"[CI] Remove periodic stable jobs from unmaintained branch","message":"[CI] Remove periodic stable jobs from unmaintained branch\n\nPeriodic stable jobs are meant for stable branches and not for\nunmaintained branches. perioid-weekly is enough (if there is\nanyone, who checks its result) to keep. This patch removes all\nperiodic-stable jobs.\n\nThis patch also removes networking-sfc from required projects in\nzuul.yaml as unmaintained/zed and older branches do not exist\nanymore for this repository. Furthermore, in this branch we have\nto cap as well networking-sfc to avoid incompatibilities with\ncurrent neutron version.\n\nLower constraints are not used and tested anymore in most of the\nOpenStack projects so it doesn\u0027t make sense to maintain it here,\nhence the job is removed and the lower-constraints.txt as well.\n\nAlso include this patch to unblock the gate: Cap setuptools \u003c71.0.0\n\npy39 jobs (on ubuntu-focal) started to fail (see the relevant github\nissue [1]) due to recent virtualenv release (20.26.4; which bundles\nsetuptools).\n\nsetuptools is bundled in virtualenv, so it has to be capped via the\nvirtualenv package. tox also needed to be capped (\u003c4) as gate uses\ntox 3.28.0, but with capping virtualenv we pull in latest tox as well,\nwhich would cause other errors.\n\n[1] https://github.com/pypa/setuptools/issues/4483\n\nConflicts:\n  .zuul.yaml\n\nChange-Id: I085f3343e1466f43e4318055e252347000397f0f\n(cherry picked from commit 79b2d924c9baeb4d824882c0473f0ca0060832cb)\n(cherry picked from commit 4570f681ec7873d854dbca32508008c292315491)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e93b33d044ce91eb1afb9f6c491df8cd0bae3844"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/e93b33d044ce91eb1afb9f6c491df8cd0bae3844"}]},"branch":"refs/heads/unmaintained/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
