)]}'
{"id":"openstack%2Fnetworking-bagpipe~943826","triplet_id":"openstack%2Fnetworking-bagpipe~unmaintained%2Fxena~I085f3343e1466f43e4318055e252347000397f0f","project":"openstack/networking-bagpipe","branch":"unmaintained/xena","topic":"remove-periodic-stable-from-unmaintained","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-03-21 18:34:32.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I085f3343e1466f43e4318055e252347000397f0f","subject":"[CI] Remove periodic stable jobs from unmaintained branch","status":"ABANDONED","created":"2025-03-07 21:10:10.000000000","updated":"2025-03-21 18:34:32.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"83b9f81584dec1dbc57a6649d47eddbf4491391b","_number":943826,"virtual_id_number":943826,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-03-09 11:49:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2025-03-10 08:13:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-07 22:21:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-10 08:12:59.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2025-03-10 08:13:06.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-03-21 18:34:32.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"3f7f3409c4efd897557c6a1aec19b42c64202a64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-07 21:10:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f39fa91f24c158ab2c56dc0ebb70965a631379e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 22:21:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf2579a39724408ebe3e249578530ef3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61a1a4d323e3455db286a858ab3c6734 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7203719353724eef8aa257413d2d6470 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f791d884a8254a498dee8d7f5476dc04 : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5655aea9a6be4d5d9cb3e750f55062db : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e97e51c9a344912a5dd64e9e7c4540e : SUCCESS in 4m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0755f7bec99e408e891ad367f7133f13 : SUCCESS in 2m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6893f478bca3467b8d78c96f61ed5057 : SUCCESS in 2m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6dcf47703b14e33889aa2c2a636d9f9 : SUCCESS in 7m 50s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/ace2992da32b4877bd01c95d27d755c4 : FAILURE in 40m 15s (non-voting)\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/17480f1970f74fffaa9cdc661ab36bdb : FAILURE in 13m 09s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/7cb94c9b34aa4f0183504ea518e0a474 : FAILURE in 1h 01m 10s","accounts_in_message":[],"_revision_number":1},{"id":"1a67e0c5713fce5477976d65e011226d1a32ec7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-09 10:59:53.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7ab7b51ee9dfa64eb2987f271814f049d255f6e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-09 11:49:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6646f9c805eb49f39da1059449785fb4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fa8e1aab56045e781aaf1387d73658f : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ead329a534134115afdcb26f66f9000d : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1ef562a50bc40fbaa8dd57783daee14 : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca473e5a18fc4369a1fb4d1a1e9c0d4d : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7e32912e5054568840960f7a9cb6943 : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d80bcf190d6d4aafa60b5ce06dbef800 : SUCCESS in 1m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a914f4a453274f22bd5f4d6c8d9241ea : SUCCESS in 1m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04a052898b784ce7a313b55a75e34804 : SUCCESS in 3m 42s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/f77fd8edfc5642da8dd85277b6243717 : FAILURE in 18m 18s (non-voting)\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/4263cae775934f7bb00f259ec14690a1 : FAILURE in 5m 59s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/032b803ce71d4eb08fe7566023657bef : FAILURE in 48m 24s","accounts_in_message":[],"_revision_number":2},{"id":"0501739a7c4c07d338e16cac915a84a42e8e8289","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-03-10 08:12:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"611209952ace4a98cc65eaf50420de5496f4f41a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-03-10 08:13:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3e727a095d126725321eb1818d6e1d3857f33317","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 08:57:58.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3b0d068dec1146d284c4e35dba2403aa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/862a44a062834208b367ad19376afd39 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6330152c1a744a1b8a32ee44be2f614 : SUCCESS in 8m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b66cff3870f14d4ebb1c9584a4dcf70a : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bfff158b65824bc09c3df970f4b06553 : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fb5303f90114cf999860e22b673008e : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a368bbba9aa2438f97e932463d1653f7 : SUCCESS in 2m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a7757d8ca553431fa9805e96355ffb3b : SUCCESS in 2m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/849419eb3daf430a960dbdab5a28b243 : SUCCESS in 7m 56s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/5b3b7118f4db42ee91bd039081892f2c : FAILURE in 39m 19s (non-voting)\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/0a133139082e4f089ea35fdd6492db88 : FAILURE in 6m 01s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/62e7be15d2604fab92ee69d7c4da211d : FAILURE in 30m 59s","accounts_in_message":[],"_revision_number":2},{"id":"83b9f81584dec1dbc57a6649d47eddbf4491391b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-21 18:34:32.000000000","message":"Abandoned\n\nunmaintained/xena branch of openstack/networking-bagpipe transitioned to End of Life and is about to be deleted. To be able to do that, all open patches need to be abandoned.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3e6563ffb0ad7e0ca2c3034c742898b799b857b9","revisions":{"ad34c730b70e4f39008b7c2f7366ba913fc7496c":{"kind":"REWORK","_number":1,"created":"2025-03-07 21:10:10.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/26/943826/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/26/943826/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/943826/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/943826/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/943826/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/943826/1"}}},"commit":{"parents":[{"commit":"39546f061193de8d46d6ca6141122f83bb18621a","subject":"Add required projects where necessary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/39546f061193de8d46d6ca6141122f83bb18621a"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-02-24 13:22:17.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-03-07 21:09:49.000000000","tz":60},"subject":"[CI] Remove periodic stable jobs from unmaintained branch","message":"[CI] Remove periodic stable jobs from unmaintained branch\n\nPeriodic stable jobs are meant for stable branches and not for\nunmaintained branches. perioid-weekly is enough (if there is\nanyone, who checks its result) to keep. This patch removes all\nperiodic-stable jobs.\n\nThis patch also removes networking-sfc from required projects in\nzuul.yaml as unmaintained/zed and older branches do not exist\nanymore for this repository. Furthermore, in this branch we have\nto cap as well networking-sfc to avoid incompatibilities with\ncurrent neutron version.\n\nLower constraints are not used and tested anymore in most of the\nOpenStack projects so it doesn\u0027t make sense to maintain it here,\nhence the job is removed and the lower-constraints.txt as well.\n\nAlso include this patch to unblock the gate: Cap setuptools \u003c71.0.0\n\npy39 jobs (on ubuntu-focal) started to fail (see the relevant github\nissue [1]) due to recent virtualenv release (20.26.4; which bundles\nsetuptools).\n\nsetuptools is bundled in virtualenv, so it has to be capped via the\nvirtualenv package. tox also needed to be capped (\u003c4) as gate uses\ntox 3.28.0, but with capping virtualenv we pull in latest tox as well,\nwhich would cause other errors.\n\n[1] https://github.com/pypa/setuptools/issues/4483\n\nConflicts:\n  .zuul.yaml\n\nChange-Id: I085f3343e1466f43e4318055e252347000397f0f\n(cherry picked from commit 79b2d924c9baeb4d824882c0473f0ca0060832cb)\n(cherry picked from commit 4570f681ec7873d854dbca32508008c292315491)\n(cherry picked from commit e93b33d044ce91eb1afb9f6c491df8cd0bae3844)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/ad34c730b70e4f39008b7c2f7366ba913fc7496c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/ad34c730b70e4f39008b7c2f7366ba913fc7496c"}]},"branch":"refs/heads/unmaintained/xena"},"3e6563ffb0ad7e0ca2c3034c742898b799b857b9":{"kind":"REWORK","_number":2,"created":"2025-03-09 10:59:53.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/26/943826/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/26/943826/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/943826/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/943826/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/943826/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/26/943826/2"}}},"commit":{"parents":[{"commit":"39546f061193de8d46d6ca6141122f83bb18621a","subject":"Add required projects where necessary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/39546f061193de8d46d6ca6141122f83bb18621a"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-02-24 13:22:17.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-03-09 10:59:02.000000000","tz":60},"subject":"[CI] Remove periodic stable jobs from unmaintained branch","message":"[CI] Remove periodic stable jobs from unmaintained branch\n\nPeriodic stable jobs are meant for stable branches and not for\nunmaintained branches. perioid-weekly is enough (if there is\nanyone, who checks its result) to keep. This patch removes all\nperiodic-stable jobs.\n\nThis patch also removes networking-sfc from required projects in\nzuul.yaml as unmaintained/zed and older branches do not exist\nanymore for this repository. Furthermore, in this branch we have\nto cap as well networking-sfc to avoid incompatibilities with\ncurrent neutron version.\n\nLower constraints are not used and tested anymore in most of the\nOpenStack projects so it doesn\u0027t make sense to maintain it here,\nhence the job is removed and the lower-constraints.txt as well.\n\nAlso include this patch to unblock the gate: Cap setuptools \u003c71.0.0\n\npy39 jobs (on ubuntu-focal) started to fail (see the relevant github\nissue [1]) due to recent virtualenv release (20.26.4; which bundles\nsetuptools).\n\nsetuptools is bundled in virtualenv, so it has to be capped via the\nvirtualenv package. tox also needed to be capped (\u003c4) as gate uses\ntox 3.28.0, but with capping virtualenv we pull in latest tox as well,\nwhich would cause other errors.\n\n[1] https://github.com/pypa/setuptools/issues/4483\n\nConflicts:\n  .zuul.yaml\n\nChange-Id: I085f3343e1466f43e4318055e252347000397f0f\n(cherry picked from commit 79b2d924c9baeb4d824882c0473f0ca0060832cb)\n(cherry picked from commit 4570f681ec7873d854dbca32508008c292315491)\n(cherry picked from commit e93b33d044ce91eb1afb9f6c491df8cd0bae3844)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/3e6563ffb0ad7e0ca2c3034c742898b799b857b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/3e6563ffb0ad7e0ca2c3034c742898b799b857b9"}]},"branch":"refs/heads/unmaintained/xena"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
