)]}'
{"id":"openstack%2Fnetworking-bagpipe~949030","triplet_id":"openstack%2Fnetworking-bagpipe~stable%2F2024.2~I7954e88ab755b13a36a45149bcfabb37b8e9bfac","project":"openstack/networking-bagpipe","branch":"stable/2024.2","topic":"bagpipe-stable-periodic-fix","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2025-05-09 22:37:04.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-05-09 20:16:10.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-05-09 22:37:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7954e88ab755b13a36a45149bcfabb37b8e9bfac","subject":"[CI][stable-only] Fix py311 periodic stable job for 2024.2","status":"MERGED","created":"2025-05-07 12:59:51.000000000","updated":"2025-05-09 22:38:07.000000000","submitted":"2025-05-09 22:37:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949030-bagpipe-stable-periodic-fix","meta_rev_id":"a08eb7cbef899d423b956520d3baec40193b573a","_number":949030,"virtual_id_number":949030,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-09 22:37:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-05-09 15:18:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2025-05-09 15:18:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-05-09 20:17:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-07 13:02:53.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2025-05-07 13:17:07.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-05-07 14:29:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-08 12:25:06.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"f0367c8e1a5805439320a1c30e1eb063400d6142","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-05-07 12:59:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee60834869438d8ea08d8a58af675ae6eadb5a41","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-05-07 13:02:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f57429b3a739c9c31d45376731f912a22a2292ba","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-07 13:17:07.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6aac273817b2472ff226f609f00c1caf5ccbf60f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 14:29:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a16a279c772471b85d5eb6b5d132efa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc6e2dcee86a4d36a6e35673e45cab17 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26024760d5d741fa90e726cdc491466e : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/35c6a9254e5b452da0a43a39f6f31e99 : SUCCESS in 7m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fe21dc92c48c486d8c060e41f47991ca : SUCCESS in 7m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7db8c344718546c28231426f044a38e7 : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b03aff4dc5b40aa8052936e4f9c83bf : SUCCESS in 5m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d4ae6a4b84b4898b2dca73e488e3aa7 : SUCCESS in 7m 47s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/321a5b414d654d4cbb98f44233edc61e : FAILURE in 25m 38s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/76e1ccdf2ab44628bc3e47d62c5fb44c : FAILURE in 9m 47s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/145449a495884ae79db200f7a3931cc2 : SUCCESS in 1h 27m 36s\n\nWarning:\n  Comments left for invalid file .pylintrc","accounts_in_message":[],"_revision_number":1},{"id":"ec8453b16d4d7e76d088dd826afb1847576e1d69","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-05-07 14:47:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7f04262c5705b190321ee8f9ca30134d93976dcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 16:15:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e3d67dfbca44421b9ddbd6504ab1d08\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dc0caeccb7e448eab0bd8e32ecf633c : SUCCESS in 9m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6eeccb44f944ad88fc7de70cdc36ee4 : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a954d675bdf44139b727027b16f794bd : SUCCESS in 5m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e315b77ea331408ab5c529a3d4a963fc : SUCCESS in 5m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/669f687e274d4c23a9d41d35da0d3eda : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2730698e9a1c4b01b7c86797bc669178 : SUCCESS in 8m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c77ac16eba704ffa827b2e5f1125316f : SUCCESS in 8m 07s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/595f69c7bf354b28976d194ee594b55e : FAILURE in 10m 23s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/3fccc60a1575434097ed36f6291d5321 : FAILURE in 6m 33s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/ba1f687e13af49c09f4fa65afd0356f8 : SUCCESS in 1h 22m 14s\n\nWarning:\n  Comments left for invalid file .pylintrc","accounts_in_message":[],"_revision_number":1},{"id":"b5793f6eacc357eb5986e4a5dfba3738824f2a82","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-05-08 12:25:06.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"392a431f04c87d452c87eec51abbe1fd6d8cdd09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 14:02:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3099de16dd3d47018a89a671680b0248\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da483b1bce274b4d862b9d1a84d28b96 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e0b8a24089d4902acb76a2a9e7cd670 : SUCCESS in 8m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5cf4d5acbc99478c8d6b93bffcf7a4ad : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/541ced0cf1394aa49d5411f1b28e4201 : SUCCESS in 7m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/af7d3767fb9b4d1ea686b87c5a62f320 : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f57abd2eb8e742c59b5055c8b4db58d6 : SUCCESS in 11m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5a9dafb2f49417b95505e329d4ae2c0 : SUCCESS in 4m 49s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/ff3a2702ef624ca59f6e79392ad1db2c : SUCCESS in 1h 32m 12s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/0aa5d697476843fb8bc9178f40a67bba : FAILURE in 10m 36s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/aa73c33ebce5475ca4d515e04c89e638 : SUCCESS in 1h 21m 05s\n\nWarning:\n  Comments left for invalid file .pylintrc","accounts_in_message":[],"_revision_number":1},{"id":"9c7c578b9b7e4a09078cfba3c4fdaaa631b75c6c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-05-08 14:12:28.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5295b00fbab7d8f78442ad0bf0203cdd2505848f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 14:12:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"72068a593b216f5939bd441034fd9a08e615091c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 15:50:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f57022852b3848c5ba7cddc0ebd5b7e2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9998a03b950041e8a5f8252fe25f27c2 : SUCCESS in 9m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e0f3b78b3d04b719b906367efe0d80f : SUCCESS in 7m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e49da5dbe9040159fbf92d5de7c124e : SUCCESS in 6m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e5bd32235182497bbe1faa748056bd18 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5bb219189ec449a82d3cb92e5aa95aa : SUCCESS in 9m 11s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/5bcebad982fc4696b4694a7508ff23d8 : FAILURE in 1h 37m 47s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/86ef7e5ecf5b4ffeadb134ee9dacef3a : SUCCESS in 1h 26m 08s\n\nWarning:\n  Comments left for invalid file .pylintrc","accounts_in_message":[],"_revision_number":1},{"id":"9feef73fe752c92732bba9325e4b4798b8d13217","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-05-09 14:40:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3a723d0df485907cede27d174974ab52be2f71f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"real_author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-05-09 15:18:48.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_17685\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"_revision_number":2},{"id":"310f10af5aa5732d1fcbd5fe61dbad10c12d8c11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 16:33:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3769747f356e46e5bc28f7a12f38bdc7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0164a1abd2bc42d88d2394d42d824c83 : SUCCESS in 10m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57a10f762df44e76afb7635aad931957 : SUCCESS in 10m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/28ef07bb2cd647eba2d1715a567f88d1 : SUCCESS in 9m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/87e2308bba6f41c6b2015efae3c8fc30 : SUCCESS in 10m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b14dc6a488554697b050f65a6a0d85c1 : SUCCESS in 10m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ce1b5f7adef4a99a18e7f28191936aa : SUCCESS in 9m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5da94792eea42669a164f324ca52b14 : SUCCESS in 9m 24s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/b59edb9b40ab4ef4b4e1b9345c837e61 : SUCCESS in 1h 13m 44s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/f9fe929126d34e7d99f390bbc4efb598 : FAILURE in 6m 24s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/7c0bd7d9a9884a84b0abb1d465b138b7 : FAILURE in 44m 17s\n\nWarning:\n  Comments left for invalid file .pylintrc","accounts_in_message":[],"_revision_number":2},{"id":"f1763d5a4f264cd675e8c0580d639ff38266ad27","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-05-09 20:16:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b88c2974d20d74554f1c220611aa525e9902c76b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-05-09 20:17:20.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f162cddabb6452e498cc04202d798fe6b1fdf922","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 21:36:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74c3616305ab4a0e91de86fe0ace701b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbaf99e93c6b4be3a312e73cab15ec3f : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2649021d29874e9ea96dbe1f32259c4f : SUCCESS in 6m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7f7887899c8341faa0ec9d0926cf38ee : SUCCESS in 5m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/360f1ae3060d479d9dc0308ef60e95f1 : SUCCESS in 5m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0403942a5cd2434ba7f150975e766906 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5053966437ec454f8b77cb8488a01084 : SUCCESS in 8m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/efdd2fd473a045dfbc7b5b6541e82776 : SUCCESS in 3m 42s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/0bf224e675234d7e89ad8154365a139b : SUCCESS in 1h 19m 23s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/a0d067dc28d64ac591151a1c463d1c79 : FAILURE in 6m 17s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/cd091eb887bd466982f75b98c877173e : SUCCESS in 1h 02m 38s\n\nWarning:\n  Comments left for invalid file .pylintrc","accounts_in_message":[],"_revision_number":2},{"id":"0c8d7fbdc861c252467762192b843877197bb2ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 21:37:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"07b6e0ddf78d1b6449ca820a9de6853899fa931a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 22:37:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c874a50717a4242886023a2b9358919\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d714c415a6974ad1955c26592f82a873 : SUCCESS in 8m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21fee276fb634c15bbe642dd1301645d : SUCCESS in 7m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0292d12fc8a44f71a3b9f8ea5928abac : SUCCESS in 7m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a89037b0e2a447ec9772a4ad2e1d5bc4 : SUCCESS in 7m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/171e66844aaa477ba20db3c4d8c92eaa : SUCCESS in 10m 22s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/eaf71483da6b4eb19d70c2cc7de96171 : SUCCESS in 59m 30s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/9417e78960474b4b9e1b4d0ae2b4dd1e : SUCCESS in 42m 01s\n\nWarning:\n  Comments left for invalid file .pylintrc","accounts_in_message":[],"_revision_number":2},{"id":"40d613e0c5c952c4992b2902d80e9292a52923b5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 22:37:04.000000000","message":"Change has been successfully merged\n\n1 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":2},{"id":"a08eb7cbef899d423b956520d3baec40193b573a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 22:38:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d30eff20fc5544b99e9acf0f4b45ba5e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7191bd582144253b222ad6eb6389e02 : SUCCESS in 33s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"481855b60e4633fb8da41c3775385bad14b47d1c","revisions":{"985441714d3460cc7a0c61f9748cf9ebc7a9f93c":{"kind":"REWORK","_number":1,"created":"2025-05-07 12:59:51.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/30/949030/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/30/949030/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/949030/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/949030/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/949030/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/949030/1"}}},"commit":{"parents":[{"commit":"b9fed9808ad5c4ebaecb37f30a1c627080d9d85b","subject":"Stable Only: Use 2024.2 tempest job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b9fed9808ad5c4ebaecb37f30a1c627080d9d85b"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-05-07 12:53:51.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-05-07 12:53:51.000000000","tz":120},"subject":"[CI][stable-only] Fix py311 periodic stable job for 2024.2","message":"[CI][stable-only] Fix py311 periodic stable job for 2024.2\n\nSince bagpipe uses periodic-stable-jobs-neutron job template and it has\npy311 configured [1] we need to add it explicitly in bagpipe\u0027s periodic\nstable job list with using bagpipe\u0027s required projects included.\n\nThis is needed only for this branch as the other jobs are already\ndefined.\n\n[1] https://opendev.org/openstack/openstack-zuul-jobs/src/commit/f1140fe7bab6ac124ebd4d68dee832d26462faec/zuul.d/project-templates.yaml#L2934\n\nChange-Id: I7954e88ab755b13a36a45149bcfabb37b8e9bfac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/985441714d3460cc7a0c61f9748cf9ebc7a9f93c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/985441714d3460cc7a0c61f9748cf9ebc7a9f93c"}]},"branch":"refs/heads/stable/2024.2"},"481855b60e4633fb8da41c3775385bad14b47d1c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-09 15:18:48.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"real_uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/30/949030/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/30/949030/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/949030/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/949030/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/949030/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/30/949030/2"}}},"commit":{"parents":[{"commit":"c3b3c66aa4d5a2d3cdd31d357d193a8b74d2e00a","subject":"[CI][stable-only] Use correct version of tempest job on gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/c3b3c66aa4d5a2d3cdd31d357d193a8b74d2e00a"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-05-07 12:53:51.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-05-09 15:18:48.000000000","tz":0},"subject":"[CI][stable-only] Fix py311 periodic stable job for 2024.2","message":"[CI][stable-only] Fix py311 periodic stable job for 2024.2\n\nSince bagpipe uses periodic-stable-jobs-neutron job template and it has\npy311 configured [1] we need to add it explicitly in bagpipe\u0027s periodic\nstable job list with using bagpipe\u0027s required projects included.\n\nThis is needed only for this branch as the other jobs are already\ndefined.\n\n[1] https://opendev.org/openstack/openstack-zuul-jobs/src/commit/f1140fe7bab6ac124ebd4d68dee832d26462faec/zuul.d/project-templates.yaml#L2934\n\nChange-Id: I7954e88ab755b13a36a45149bcfabb37b8e9bfac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/481855b60e4633fb8da41c3775385bad14b47d1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/481855b60e4633fb8da41c3775385bad14b47d1c"}]},"branch":"refs/heads/stable/2024.2","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
