)]}'
{"id":"openstack%2Fnetworking-bagpipe~960497","triplet_id":"openstack%2Fnetworking-bagpipe~stable%2F2025.2~Ic15a7c59cd194ff692389a2249c715a3895c0fb7","project":"openstack/networking-bagpipe","branch":"stable/2025.2","topic":"create-2025.2","hashtags":[],"change_id":"Ic15a7c59cd194ff692389a2249c715a3895c0fb7","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.2","status":"MERGED","created":"2025-09-11 12:31:24.000000000","updated":"2025-09-16 08:14:06.000000000","submitted":"2025-09-16 08:12:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"960497-create-2025.2","meta_rev_id":"31a796e447acd4a6641d9371b2100eb8c67481b9","_number":960497,"virtual_id_number":960497,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-16 08:12:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-09-16 07:27:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-09-16 07:27:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-11 12:59:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-16 07:27:41.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"94e17349992c41f60beefaeb1a79c878a444e89e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2025-09-11 12:31:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e4cea6b2e7ad87795cca13ccb8e121c318fe1fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 12:59:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/233e33379e2244b39324302140f6b61a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c2b982fea0c46a2918784b3a1f482ac : FAILURE in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/16ac61272bba4c4eb722556b3026c31f : SUCCESS in 6m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3433582cab6940da87e2b1f34d1be10f : SUCCESS in 7m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/45f2835ae4f24ff9bd83f748e455b9d5 : SUCCESS in 10m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f8707e9824c49a2968d809d4b798bb0 : SUCCESS in 5m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d97f0367ae14126b3670590a605cc73 : SUCCESS in 2m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/199d4497c1724c58bc3cabb84eaac06d : SUCCESS in 7m 29s\n\nWarning:\n  Comments left for invalid file networking_bagpipe/agent/bagpipe_bgp_agent.py\n  Comments left for invalid file networking_bagpipe/agent/bgpvpn/agent_extension.py\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/cli/rest_attach.py\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/common/config.py\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/common/utils.py\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/engine/bgp_peer_worker.py\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/engine/exabgp_peer_worker.py\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/engine/route_table_manager.py\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/fakerr.py\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/vpn/evpn/__init__.py\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/vpn/evpn/ovs.py\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/vpn/identifier_allocators.py\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/vpn/ipvpn/__init__.py\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/vpn/ipvpn/mpls_linux_dataplane.py\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/vpn/ipvpn/mpls_ovs_dataplane.py\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/vpn/manager.py\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/vpn/vpn_instance.py\n  Comments left for invalid file networking_bagpipe/driver/mech_bagpipe.py","accounts_in_message":[],"_revision_number":1},{"id":"76d58ca46e1c97133a7599543b41ac1c5e0cac1b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-09-16 07:27:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9ae1b5b4f9acd6e77b152f06262327bdc14d8673","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 07:51:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/532f313401b94c91b80fdc5cb3cf8cee\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ef6f86a96c64acabd693a39b10fcd00 : SUCCESS in 6m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/351f6210cbc34479966b557bc4f16492 : SUCCESS in 7m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a5ef7dc63d1a4db685c96df3962685c2 : SUCCESS in 7m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/500f37cfd2d54b1ab802c3a9f01f883a : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dc6aa22645a44139da405af5abc7bac : SUCCESS in 9m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f006d2f545a443f2903de0953d80dccf : SUCCESS in 5m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9de18dd49b84df7bcc17558dcb46a04 : SUCCESS in 8m 39s","accounts_in_message":[],"_revision_number":1},{"id":"7598d90058b959650f73a8603a8572aba95f5672","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 07:51:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"80fb1a315db9cb2f2b0d37923b83f5d9dd14da0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 08:12:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13c5c39534184c11972e6f641b31ddc8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4b636116347431fb0ea180159e55cbb : SUCCESS in 7m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15d18345c23d4f299757e7a26a38e5c6 : SUCCESS in 8m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e22150eab12e4705a1ee9fafc98ac64c : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f974e4566df42a8add170cf6efeab2f : SUCCESS in 10m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7951f61159545aa974656a15dab3c45 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":1},{"id":"73ad964636dd0ce6b79435b9e675cf885e5de765","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 08:12:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"31a796e447acd4a6641d9371b2100eb8c67481b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 08:14:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e123c0c271dd43c191583101ae976d9b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fda7eed8af94c27b73854dae05d69a2 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8bea5b637f864e89bd4f042bce96dfeb : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b19dbab817498e35601e737af3919ab52fff4481","revisions":{"b19dbab817498e35601e737af3919ab52fff4481":{"kind":"REWORK","_number":1,"created":"2025-09-11 12:31:24.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/97/960497/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/97/960497/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/97/960497/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/97/960497/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/97/960497/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/97/960497/1"}}},"commit":{"parents":[{"commit":"6d3e50a296de54a96d9ad0b9231d76e8c4f36b3f","subject":"Update .gitreview for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/6d3e50a296de54a96d9ad0b9231d76e8c4f36b3f"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-09-11 12:31:23.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-09-11 12:31:23.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.2","message":"Update TOX_CONSTRAINTS_FILE for stable/2025.2\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/2025.2 branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Ic15a7c59cd194ff692389a2249c715a3895c0fb7\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b19dbab817498e35601e737af3919ab52fff4481"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/b19dbab817498e35601e737af3919ab52fff4481"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
