)]}'
{"id":"openstack%2Fnetworking-bagpipe~977634","triplet_id":"openstack%2Fnetworking-bagpipe~master~Iee7af0050cb6deb3d176d824d35c570bd35ba6d3","project":"openstack/networking-bagpipe","branch":"master","topic":"nodeset","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-03-02 17:23:53.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-02-27 18:10:24.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Iee7af0050cb6deb3d176d824d35c570bd35ba6d3","subject":"CI: Fix failing jobs due to nodeset and req bump","status":"MERGED","created":"2026-02-23 13:42:24.000000000","updated":"2026-03-02 17:38:41.000000000","submitted":"2026-03-02 17:23:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"977634-nodeset","meta_rev_id":"d6853d84a5f340ab8caf51a5c2e12e1f1c76b4fe","_number":977634,"virtual_id_number":977634,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-02 17:23:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-03-02 15:06:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-03-02 15:05:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-02-27 18:10:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-03-02 15:05:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-23 15:11:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-24 22:37:27.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-02-27 15:38:31.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-03-02 15:05:59.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2026-03-02 15:06:13.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"39bd237f23a6aef6feebb944a3534fe2c48df525","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-02-23 13:42:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4c9304be788402c03b344ac7fab94c817916414","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 15:11:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1814c884f5784c038f4fce595698df03\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65c4e4fc5fcd4c18bf3fee12460c92bc : FAILURE in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c8eb190344546328b7206f7a05635c7 : FAILURE in 3m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/395461a2097b4581b474d93bbca77bf8 : FAILURE in 5m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/77e732e8080e4ab8adb862fae7503b43 : FAILURE in 7m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7fe67755a2bf420a9f9bc29f605a9047 : FAILURE in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/735917f2d1234a99ac290c64fb793981 : FAILURE in 4m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d65a60e22614a859be48fa4e6237875 : FAILURE in 4m 57s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/4eccd0bec0464d458f064c0d497e9bfd : FAILURE in 35m 59s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/64d25f14056649af97a1ce92b92c7605 : SUCCESS in 1h 23m 34s\n\nWarning:\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/engine/exa.py\n  Comments left for invalid file networking_bagpipe/bagpipe_bgp/engine/exabgp_peer_worker.py","accounts_in_message":[],"_revision_number":1},{"id":"e383eead61d8df0b1629018284edb43412dc6b46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-02-24 10:18:47.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2816877b52446a875825ae22efd8f80b4811da15","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-02-24 10:19:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"975a4365fd50899c78fac58158ab00a69a608438","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-02-24 11:42:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6fce93096bf998bbd78ab2864ca531408de97d23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 13:16:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0930f20b98d44e2be88045011283093\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf632e2028d34f9d93aa4aa302712c81 : SUCCESS in 7m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f21d76da55f4ba08eb8b577bb39e98f : SUCCESS in 4m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e31ce32aae1d41a18943ebf6dd9e171f : SUCCESS in 6m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e52e55093f194531b3d3a79586329047 : SUCCESS in 5m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c53d384a488c463ca7c7413f093cb2cf : FAILURE in 6m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b23cb4433e5410faf077aae38f31d74 : SUCCESS in 10m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/13b1442d48f849e9964c4608623cff1b : SUCCESS in 3m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72865fe9bde44b4582916d349d06f829 : SUCCESS in 9m 53s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/a3978a86eee04ba48bbaa11cc526e61a : SUCCESS in 1h 31m 32s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/9e0ea4d59b314b389d07358383d06de2 : FAILURE in 41m 17s","accounts_in_message":[],"_revision_number":3},{"id":"e2a2e5a8e1da4dadc1c5af944fa1f9e27dcac248","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-02-24 14:00:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bfff71d9132b88a4ccb8ae94de696ec3d4628474","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 15:44:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68667074635e4e03a7fb8f110b223870\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89ed6f281fb5439497a827942cd34a0b : SUCCESS in 6m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/04db11035f0948c98d84a0b33db526df : SUCCESS in 5m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e3dfd69d0256436d8a1bb9ef7280a37d : SUCCESS in 7m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/64e23ac3be984a2ca38fb3001278dd5b : SUCCESS in 3m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6fd81193efeb45efabbdfa6e070ba928 : FAILURE in 3m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c20fb1230d0744f9968be5f110c8ef40 : SUCCESS in 8m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b3c8ace91a7c47f491ae68b4691c9a17 : SUCCESS in 2m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c699f53c6a374521a4d05b573e88bc28 : SUCCESS in 6m 24s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/4833713bceb34983828aefb9279c2b68 : SUCCESS in 55m 56s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/ae30240f998d4147be2fe4037e145e4c : SUCCESS in 1h 41m 57s","accounts_in_message":[],"_revision_number":3},{"id":"cf9284eaec2bca775e1b731162506bf4f657b4ba","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-24 22:37:27.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8bb780fc1d5674b00d00411c511c3bbbceb2ec84","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-02-26 10:04:39.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"9aff65fe454ce19edb536e154fda2ce3ca670d85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-02-27 08:22:08.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"9a7bf63815825bbf9fd0d720c93cf01776b97db4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-02-27 08:24:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"535748c88c8d371aae49faec39bddc5fb52f70c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 09:51:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c0f15ddf90e44738b405640763b628e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28be6f94936d4ff091aabb1ff3e5bb0b : SUCCESS in 6m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a93ae9a87f74a179cb404eb2798ea16 : SUCCESS in 6m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1145196f03824ce086a93ab08555a7fb : SUCCESS in 7m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/07961e31aa7b4c4d9d6ca3da6f80d1f5 : SUCCESS in 4m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e3f5b93eb1564e13b92d14899240a9a1 : FAILURE in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c97077a1dac74764b9922cbf84b6eb16 : SUCCESS in 11m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f22c9c06ee244853a0f5cc88a40f3ee1 : SUCCESS in 3m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2147edac10924de29c67ca7350edcca3 : SUCCESS in 7m 32s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/be77abd1882448f184d1f65c9e0aa3ce : SUCCESS in 1h 20m 59s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/0a03f1df96504baea504fcbd04768f32 : SUCCESS in 1h 27m 43s","accounts_in_message":[],"_revision_number":4},{"id":"8fcc461cc49a1c02ffe4a73c3500c98f436451d6","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-27 15:38:31.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d687d20aad22d7a925fbab4692e06c30dac07793","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-02-27 15:52:19.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"333e2277a6a0d726e38a8cef2fd1cb74826bd5f7","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-02-27 15:53:38.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"967444e38ff41644b47c0417d9dd9d7620abae1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 17:30:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc4645a870464b4cbb9ffd1efe0aea5a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb0b35a82f034566bf3bed43548cbb8f : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4db972639fd64ee0bb5e1397b57d7842 : SUCCESS in 8m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b0032b9ef57846d38074e73850aa5808 : SUCCESS in 10m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/60b563cc415b4da78e445b498726036d : SUCCESS in 4m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1d5e4db895df478bbb5dfff75c4ce40b : FAILURE in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a988fadad6604883a2af13b38744b630 : SUCCESS in 5m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/84f2bad993b24f94988c78bfb4764e2a : SUCCESS in 2m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5701ac7a9d5d4c0aae28b186d43ea8a3 : SUCCESS in 8m 35s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/95528578cd934acbba52cf54c6f37267 : SUCCESS in 1h 35m 20s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/e50a6b1fcd98445a905c6ebf96fe3eef : SUCCESS in 1h 17m 40s","accounts_in_message":[],"_revision_number":5},{"id":"732c2c38da48c4b1e87174757993e2e6fe5e1d88","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-02-27 18:10:24.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"f74c06ae568a404e3269b2e6245cc3493d6b623e","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-03-02 15:05:59.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"ce87e9480c2e7f4d2417e5a4707e747f3aaf8638","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-03-02 15:06:13.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"3c075ab13ee5754e5be732804e609eecad303503","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 15:07:58.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"1845810b9d3ad27ecff1e77bca651a211f5a1716","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 17:23:53.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9a0083589c140ceb5570f6a9c968895\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a7620dcbce64fb58be20e3ed21bdb88 : SUCCESS in 6m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/be7ff914cf9c4b0e95b752a3aa3f58d7 : SUCCESS in 7m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9f869279958448d09a559d7df59390b4 : SUCCESS in 4m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aee0bc1f1e9142f99567764c7176e169 : SUCCESS in 7m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0949f0256ca648aeacab847ab3d7db64 : SUCCESS in 10m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/030e013f0b9c4b23b83e134fd5fa677f : SUCCESS in 4m 07s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/e7c0589deca44110ae8c17ff430fe273 : SUCCESS in 1h 47m 15s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/3aea913347c040ecab68ded228adfd4e : SUCCESS in 1h 28m 30s","accounts_in_message":[],"_revision_number":5},{"id":"ef692d7972acd1389cd13fa9ed1932fd2f19c754","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 17:23:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"d6853d84a5f340ab8caf51a5c2e12e1f1c76b4fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 17:38:41.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90bd1d2c57e642cc9fc7cc1d6f52a410\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed0d6715687e41a5bd2e83b65db42ac3 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"3ee62a379a81fa9f01e3282fc7802fef54a33f1a","revisions":{"3cb112d04dd975da9ba6e9a3ed35150d4283bbaa":{"kind":"REWORK","_number":1,"created":"2026-02-23 13:42:24.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/34/977634/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/34/977634/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/34/977634/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/34/977634/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/34/977634/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/34/977634/1"}}},"commit":{"parents":[{"commit":"54ddcb0721993aa9e5c13e0eeb097d6b9b8bd4b1","subject":"Migrate setup configuration to pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/54ddcb0721993aa9e5c13e0eeb097d6b9b8bd4b1"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-02-23 13:40:52.000000000","tz":60},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-02-23 13:40:52.000000000","tz":60},"subject":"CI: change networking-bagpipe-tempest job nodeset","message":"CI: change networking-bagpipe-tempest job nodeset\n\nFor some reason the nodeset for networking-bagpipe-tempest\nremained openstack-single-node-jammy, it is time\nto change it to openstack-single-node-noble\n\nChange-Id: Iee7af0050cb6deb3d176d824d35c570bd35ba6d3\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/3cb112d04dd975da9ba6e9a3ed35150d4283bbaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/3cb112d04dd975da9ba6e9a3ed35150d4283bbaa"}]},"branch":"refs/heads/master"},"af5e3187f90f1126ded7e5de1c5a96f907dafc4e":{"kind":"REWORK","_number":2,"created":"2026-02-24 10:18:47.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/34/977634/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/34/977634/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/34/977634/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/34/977634/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/34/977634/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/34/977634/2"}}},"commit":{"parents":[{"commit":"54ddcb0721993aa9e5c13e0eeb097d6b9b8bd4b1","subject":"Migrate setup configuration to pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/54ddcb0721993aa9e5c13e0eeb097d6b9b8bd4b1"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-02-23 13:40:52.000000000","tz":60},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-02-24 10:15:30.000000000","tz":60},"subject":"CI: Fix failing jobs due to nodeset and req bump","message":"CI: Fix failing jobs due to nodeset and req bump\n\nFor some reason the nodeset for networking-bagpipe-tempest\nremained openstack-single-node-jammy, it is time\nto change it to openstack-single-node-noble\n\nLatest requirement bum (see [1]) pulled in exabgp 5.0.1,\nand to make bagpipe code work with the reorganized code\nsome minor changes are necessary.\n\n[1]: https://review.opendev.org/c/openstack/requirements/+/965873\nChange-Id: Iee7af0050cb6deb3d176d824d35c570bd35ba6d3\nAssisted-By: claude-haiku-4.5\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/af5e3187f90f1126ded7e5de1c5a96f907dafc4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/af5e3187f90f1126ded7e5de1c5a96f907dafc4e"}]},"branch":"refs/heads/master"},"550b8262410c7f9be7d7dccc38f980a50c4cafcf":{"kind":"REWORK","_number":3,"created":"2026-02-24 11:42:01.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/34/977634/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/34/977634/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/34/977634/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/34/977634/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/34/977634/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/34/977634/3"}}},"commit":{"parents":[{"commit":"54ddcb0721993aa9e5c13e0eeb097d6b9b8bd4b1","subject":"Migrate setup configuration to pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/54ddcb0721993aa9e5c13e0eeb097d6b9b8bd4b1"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-02-23 13:40:52.000000000","tz":60},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-02-24 11:41:34.000000000","tz":60},"subject":"CI: Fix failing jobs due to nodeset and req bump","message":"CI: Fix failing jobs due to nodeset and req bump\n\nFor some reason the nodeset for networking-bagpipe-tempest\nremained openstack-single-node-jammy, it is time\nto change it to openstack-single-node-noble\n\nLatest requirement bum (see [1]) pulled in exabgp 5.0.1,\nand to make bagpipe code work with the reorganized code\nsome minor changes are necessary.\n\n[1]: https://review.opendev.org/c/openstack/requirements/+/965873\nChange-Id: Iee7af0050cb6deb3d176d824d35c570bd35ba6d3\nAssisted-By: claude-haiku-4.5\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/550b8262410c7f9be7d7dccc38f980a50c4cafcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/550b8262410c7f9be7d7dccc38f980a50c4cafcf"}]},"branch":"refs/heads/master"},"d1b189f5c1e01a5c1253fa7b0e8c8842ce2e8d70":{"kind":"REWORK","_number":4,"created":"2026-02-27 08:22:08.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/34/977634/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/34/977634/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/34/977634/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/34/977634/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/34/977634/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/34/977634/4"}}},"commit":{"parents":[{"commit":"54ddcb0721993aa9e5c13e0eeb097d6b9b8bd4b1","subject":"Migrate setup configuration to pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/54ddcb0721993aa9e5c13e0eeb097d6b9b8bd4b1"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-02-23 13:40:52.000000000","tz":60},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-02-27 08:21:53.000000000","tz":60},"subject":"CI: Fix failing jobs due to nodeset and req bump","message":"CI: Fix failing jobs due to nodeset and req bump\n\nFor some reason the nodeset for networking-bagpipe-tempest\nremained openstack-single-node-jammy, it is time\nto change it to openstack-single-node-noble\n\nLatest requirement bum (see [1]) pulled in exabgp 5.0.1,\nand to make bagpipe code work with the reorganized code\nsome minor changes are necessary.\n\n[1]: https://review.opendev.org/c/openstack/requirements/+/965873\nChange-Id: Iee7af0050cb6deb3d176d824d35c570bd35ba6d3\nAssisted-By: claude-haiku-4.5\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/d1b189f5c1e01a5c1253fa7b0e8c8842ce2e8d70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/d1b189f5c1e01a5c1253fa7b0e8c8842ce2e8d70"}]},"branch":"refs/heads/master"},"3ee62a379a81fa9f01e3282fc7802fef54a33f1a":{"kind":"REWORK","_number":5,"created":"2026-02-27 15:52:19.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/34/977634/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/34/977634/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/34/977634/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/34/977634/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/34/977634/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/34/977634/5"}}},"commit":{"parents":[{"commit":"54ddcb0721993aa9e5c13e0eeb097d6b9b8bd4b1","subject":"Migrate setup configuration to pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/54ddcb0721993aa9e5c13e0eeb097d6b9b8bd4b1"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-02-23 13:40:52.000000000","tz":60},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-02-27 15:51:29.000000000","tz":60},"subject":"CI: Fix failing jobs due to nodeset and req bump","message":"CI: Fix failing jobs due to nodeset and req bump\n\nFor some reason the nodeset for networking-bagpipe-tempest\nremained openstack-single-node-jammy, it is time\nto change it to openstack-single-node-noble\n\nLatest requirement bum (see [1]) pulled in exabgp 5.0.1,\nand to make bagpipe code work with the reorganized code\nsome minor changes are necessary.\n\n[1]: https://review.opendev.org/c/openstack/requirements/+/965873\nChange-Id: Iee7af0050cb6deb3d176d824d35c570bd35ba6d3\nAssisted-By: claude-haiku-4.5\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/3ee62a379a81fa9f01e3282fc7802fef54a33f1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/3ee62a379a81fa9f01e3282fc7802fef54a33f1a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
