)]}'
{"id":"openstack%2Fnetworking-bagpipe~990668","triplet_id":"openstack%2Fnetworking-bagpipe~unmaintained%2F2024.1~I08f896589ac806f43a14f7cae78c95ff134de31b","project":"openstack/networking-bagpipe","branch":"unmaintained/2024.1","topic":"networking-bagpipe-periodic-cleanup","attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-05-29 16:58:25.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I08f896589ac806f43a14f7cae78c95ff134de31b","subject":"[CI][stable-only] Clean up periodic jobs","status":"NEW","created":"2026-05-29 15:30:49.000000000","updated":"2026-06-04 20:05:28.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7c842adf0cd3207eb804ec7ddad672172af1c3c0","_number":990668,"virtual_id_number":990668,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-05-29 16:58:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-29 16:58:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-04 20:05:28.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"}],"messages":[{"id":"01aa1828513790dcf9785301d107ca816b14bef6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-05-29 15:30:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a375c6335ef3252f857ea88493159aec3898b5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 16:58:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ee50492960b4cd284ec2719944a64de\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/729bbc409cb147a5bd1751b111494b1e : FAILURE in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d7451bdf57243ee9c8bbd5443b19097 : SUCCESS in 7m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b714da5228e47eba526c8970811d895 : FAILURE in 5m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2399f00b6d524ba282319ba03db1631d : FAILURE in 6m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5ec10d2457334f78bdb1082814d8e6b0 : FAILURE in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef756e8ef551461db4da466b0d55452d : FAILURE in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e42569f9f2ab42828cca68dbf06c85ec : SUCCESS in 6m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59ef432b93f2492e9c4485777db0481d : FAILURE in 7m 11s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-1 https://zuul.opendev.org/t/openstack/build/6ddcc725d91f411a824b710cca1422fc : SUCCESS in 1h 02m 46s\n- networking-bagpipe-dsvm-fullstack https://zuul.opendev.org/t/openstack/build/e5fc91ea7e9a4b11b4e12fee851c45a3 : FAILURE in 16m 11s (non-voting)\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/0c9f9a59f9e84e84a5b3c15032fedc73 : SUCCESS in 1h 19m 07s\n\nWarning:\n  Comments left for invalid file .pylintrc","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"360f1dec782f30101439fccb342a0889f9ee6198","revisions":{"360f1dec782f30101439fccb342a0889f9ee6198":{"kind":"REWORK","_number":1,"created":"2026-05-29 15:30:49.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/68/990668/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/68/990668/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/68/990668/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/68/990668/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/68/990668/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/68/990668/1"}}},"commit":{"parents":[{"commit":"39167cd7f8cee209273d8077057b496d4796ee6e","subject":"Update .gitreview for unmaintained/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/39167cd7f8cee209273d8077057b496d4796ee6e"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2026-05-29 15:28:45.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2026-05-29 15:30:12.000000000","tz":120},"subject":"[CI][stable-only] Clean up periodic jobs","message":"[CI][stable-only] Clean up periodic jobs\n\n2024.1 branch is unmaintained, no need to run daily periodic-stable\njobs anymore. Neither some of the weekly periodic jobs look like\nunnecessary on unmaintained.\n\nChange-Id: I08f896589ac806f43a14f7cae78c95ff134de31b\nSigned-off-by: Előd Illés \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/360f1dec782f30101439fccb342a0889f9ee6198"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/360f1dec782f30101439fccb342a0889f9ee6198"}]},"branch":"refs/heads/unmaintained/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
