)]}'
{"id":"openstack%2Fnetworking-bagpipe~991347","triplet_id":"openstack%2Fnetworking-bagpipe~master~Ib6647e231dc4cda430f42f856b92999d0e0c3d83","project":"openstack/networking-bagpipe","branch":"master","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-06-04 19:27:23.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"Ib6647e231dc4cda430f42f856b92999d0e0c3d83","subject":"Update jobs based on testing runtime for 2026.2","status":"MERGED","created":"2026-06-03 13:26:14.000000000","updated":"2026-06-04 22:17:00.000000000","submitted":"2026-06-04 22:15:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991347","meta_rev_id":"fb56a55fcf8bb5febf35e9741d10e6006e040b90","_number":991347,"virtual_id_number":991347,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-04 22:15:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-06-03 13:26:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-06-04 09:19:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-06-04 09:19:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-03 13:26:47.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-06-03 15:12:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-04 09:19:11.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-06-04 12:17:01.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"afbeff0b8ffe1d71d236aee1540ee50eed1e5d5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-03 13:26:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ccf57853943440b9952e2da33a516be36ddd38c7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-03 13:26:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d8a489958b8911430b2445d7434d88629ed72a51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 15:12:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/418850eb33c54263962f5712e08e64f5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b778374a18b9428188001a37b50e08d7 : SUCCESS in 7m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a87e56b0e3df42dcbbc25155bef9fb30 : SUCCESS in 7m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/72b51f6b824f406aaaf32a5de9d2d9df : SUCCESS in 7m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7f32ba181bc147f88c8e32c7b8db5f33 : FAILURE in 7m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1664c100c9b4b1d9cde6d108d8bf305 : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4693dcff0e704fb9acca01eb21825acd : SUCCESS in 5m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09b8c132878045ecbebb7785f80f4d07 : SUCCESS in 4m 52s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/d7b769e2bc7c474e83e114d74dfb28ad : FAILURE in 1h 40m 44s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/dfa8a50d985049e88c005e476f6d5d6e : SUCCESS in 1h 18m 42s","accounts_in_message":[],"_revision_number":1},{"id":"329261fce7a4252aa96da3e73a9ed6689649cd3a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-03 15:22:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"df2adde8edc12f6fd1172960cc7a00f5225bf8b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 17:46:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15743af5268e46e0aebf331e94e05f34\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c14710e3b53343aab17ff41f390928be : SUCCESS in 7m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/61ded15aa6604dafb83ff5df8655bc46 : SUCCESS in 8m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f0e614f219e7492a95a2a50991f221b8 : SUCCESS in 5m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1109a19853f44e1c8ba1a7d85d3f6464 : FAILURE in 6m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7e506893c8b4b738608d89f64d494e1 : SUCCESS in 10m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e22938fdabc4955aedb50812099d5ca : SUCCESS in 6m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d268960cdd0345d2a2e122cb63c023ff : SUCCESS in 5m 06s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/f489cefdd8dc45daaa4cc547c0b87f40 : SUCCESS in 2h 05m 46s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/dd5cd5948f5848e3989c849c9e77b6dc : SUCCESS in 1h 10m 43s","accounts_in_message":[],"_revision_number":1},{"id":"14a343f637dd5415e66294d3182ee118e5a77b1c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-06-04 09:19:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"64ea0166d8a75936bf0fc502653afc0739908f84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 09:19:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"87a5039a1d527be897dd03f3c122e849d669357c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 11:12:04.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c8628063c0b43c185e764b4154128a5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cceac7ddd72b497b9784d5fea9acf18e : SUCCESS in 6m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/072ff49bb18942ab8db123947c9eb4bc : SUCCESS in 5m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4da175499acf4272836664a876134648 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b979b2e8beb4ac39341535781624db7 : SUCCESS in 10m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c82c21b5a35b4752a46411aa2a213c92 : SUCCESS in 5m 55s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/a16782410944464487dae3caae4df544 : SUCCESS in 1h 48m 25s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/c026b1509a4345759f973c6216c3df2e : FAILURE in 49m 59s","accounts_in_message":[],"_revision_number":1},{"id":"d584b8a0b98b92ffaab0d625cdc183de084ed61a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-04 12:17:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d2a0cad47f7cd58da0bcc714ecb487d628997daf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 14:03:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76b785601e094164896f601691713686\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1ded6a4d88145d3a40ea614764261b0 : SUCCESS in 7m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/726846e1b26f4ded86b166529302d7a0 : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/03ea970e86ec42979e7d5470f8836d8a : SUCCESS in 5m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2b099691223d4168b85641ea00441213 : FAILURE in 5m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5cf6fa109784e31a1da369f578d5cf9 : SUCCESS in 4m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a0b55160902402983dc8e4ff32294d6 : SUCCESS in 2m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ac940a893124273b2aa68027c57f3f0 : SUCCESS in 4m 13s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/467ccc8fcd7d4ded9908bf28752b0cc1 : SUCCESS in 1h 22m 37s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/a89afa46c8eb4882bab6a43a7b1fcd76 : SUCCESS in 1h 35m 29s","accounts_in_message":[],"_revision_number":1},{"id":"c126e5c2769107981053a9ec25f0cce02db1a167","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 14:04:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e403bc535d7b441adc10a4b2546cdec2deb3f411","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 16:03:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/35c9aedd3b284aa48eaf29198b145090\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0831dce5c5b744159f610b84f9cf2dc6 : SUCCESS in 7m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9ad9f101b8e74b009ab1ff10835db222 : SUCCESS in 6m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a216c785b3b94b3da807d74550fb41e2 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d1ff7adc76244f5bee08f9ae02afb7a : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d6307201acb424683984ab4d9b15b8b : SUCCESS in 5m 49s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/a506edf8503b49ff9130df9a11910c68 : SUCCESS in 1h 23m 16s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/f4a35f7278594e61810c0b4f1257d282 : FAILURE in 1h 55m 00s","accounts_in_message":[],"_revision_number":1},{"id":"d106f4c3aa5026c25d5e09398150f60e6fe27ec3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-04 19:27:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"180609460e5b9093589a7bc1df44c352f2d69e03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 20:27:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27211149a8534c149db6609e99cb22e6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a57680b3f19d4008bca1adfb064958c0 : SUCCESS in 7m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f5615dddf04f4cfa99e3802b034ccb91 : SUCCESS in 5m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7d91ef65aaf94ef392b35f0468efb99c : SUCCESS in 3m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7d58fde87e32428ab61afaa155365ec2 : FAILURE in 3m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/280bb641d2a34bd39f7b455626ce9fe1 : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eeb75d8358f743a78bcd5bda1aec38ab : SUCCESS in 5m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89c5dd609854491990f01b421334f4c7 : SUCCESS in 8m 50s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/feabe4c6d36e43b5b00c96bc99938f99 : SUCCESS in 58m 12s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/573550929f8a49aab8d262d5adbdbe13 : SUCCESS in 55m 36s","accounts_in_message":[],"_revision_number":1},{"id":"ce6fa5203a37ee9e8064bf53b28c9f935e0d5103","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 20:28:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ee81f37c4725658ad4bb48f6407a209b267852fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 22:15:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a80b7a7800e4460c8fe1c9830ba9b8b2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41ecd552831e49fbb04b5eb05e946792 : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7796f79830184307818f6a1b764b4a55 : SUCCESS in 5m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/96a188bd7b9744d4a415bd3dc3a41a19 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62af381b76fd41399720a983986388b4 : SUCCESS in 9m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f7553281f5c4d9694e2d1bce3338c1c : SUCCESS in 4m 58s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/e5b798ee2d5847d78d7fa2c8d3e27074 : SUCCESS in 1h 41m 44s\n- networking-bagpipe-tempest https://zuul.opendev.org/t/openstack/build/2a7ad326a78f4d32b0e38ccb0a908b11 : SUCCESS in 54m 20s","accounts_in_message":[],"_revision_number":1},{"id":"59b712567db6568a523cf32271615b96cc99d377","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 22:15:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fb56a55fcf8bb5febf35e9741d10e6006e040b90","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 22:17:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cd9b03ac0da46429392c272aedf2da0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34ab54318cb048adaa957177541b559f : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"76cf644eed10bdef18a993be2d58630c114af7c3","revisions":{"76cf644eed10bdef18a993be2d58630c114af7c3":{"kind":"REWORK","_number":1,"created":"2026-06-03 13:26:14.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/47/991347/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bagpipe","ref":"refs/changes/47/991347/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/47/991347/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/47/991347/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bagpipe refs/changes/47/991347/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bagpipe refs/changes/47/991347/1"}}},"commit":{"parents":[{"commit":"756a8d3e18518add368b125871702dff8ac8d93e","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/756a8d3e18518add368b125871702dff8ac8d93e"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-03 13:23:56.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-06-03 13:26:13.000000000","tz":-240},"subject":"Update jobs based on testing runtime for 2026.2","message":"Update jobs based on testing runtime for 2026.2\n\nBased on [0] only py311 and py313 jobs are in the testing\nruntime for 2026.2, updated relevant files accordingly.\nPy314 jobs are still non-voting.\n\nUpdate to use newer (py313) jobs in periodic and experimental\nqueues.\n\n[0] https://review.opendev.org/c/openstack/governance/+/976691\n\nChange-Id: Ib6647e231dc4cda430f42f856b92999d0e0c3d83\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/76cf644eed10bdef18a993be2d58630c114af7c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bagpipe/commit/76cf644eed10bdef18a993be2d58630c114af7c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
