)]}'
{"id":"openstack%2Fnetworking-baremetal~715411","triplet_id":"openstack%2Fnetworking-baremetal~master~I351fbf0a7937ef7b8f86e3653b0e3388e253ff32","project":"openstack/networking-baremetal","branch":"master","topic":"ci","hashtags":[],"change_id":"I351fbf0a7937ef7b8f86e3653b0e3388e253ff32","subject":"Set ram value for tinyipa job","status":"ABANDONED","created":"2020-03-27 12:19:55.000000000","updated":"2020-04-21 14:24:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6a52f2c19f9e5638ec03b484b6f8dcff77bd3ff2","_number":715411,"virtual_id_number":715411,"owner":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-04-09 09:59:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":-1,"date":"2020-04-09 13:12:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-09 09:59:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-09 13:12:36.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"944ab69f09c3983b5b1f3ccd4067e608204d3f10","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-03-27 12:19:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"279a3b8d2089a3e54ee2958c32078e7b39f4051f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 14:27:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acefadaa48ac400ea9f8b80503d17176 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9c88f6a0b7f45c494fecba061dd2d36 : SUCCESS in 6m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f433d7b467514e9088c840910b5e8708 : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75f0e85a273a4aaf8d3ac439e7e59742 : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/155586100a3142f2a2464bd5f36241b5 : SUCCESS in 8m 42s\n- networking-baremetal-multitenant-vlans https://zuul.opendev.org/t/openstack/build/d5fc6857510d44569415d971dcc8d5bd : TIMED_OUT in 2h 05m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2943add1fdd9433c9086b68ece7daa39 : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":1},{"id":"6bb6d33d0e21656bfaac720d3c1524a72041d62e","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-03-27 14:34:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ff5b1115e85a071f571b18caf964ba1ffb787e20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 16:42:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd1639555b894bd2a791fabe1d28f929 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/987689de15b5405f802d9bb103d48597 : SUCCESS in 6m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d5d00e18b49841c8a5fee08383599f94 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61cf18ff9ba14d4483d39fba153fccb4 : SUCCESS in 8m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad315949d8424ba58feae9365295cf3a : SUCCESS in 9m 13s\n- networking-baremetal-multitenant-vlans https://zuul.opendev.org/t/openstack/build/4ecd9d9f0ec2424ea3bbdbdf35bcad70 : POST_FAILURE in 2h 04m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8de24f668ff64a83b3f7089d283513cc : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":1},{"id":"d92424ba1fac65b513df49f08f81548b45b478f0","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-03-30 08:22:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"821fdc5abb182407ebab3fb392d66f65344dafaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 10:33:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e040abc37df74585a2b35acb620964f7 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cadf9cdf2963468bbaf6ffc0f9b5a4d0 : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d895a787841b4a84a7f32af8bdf707b0 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4dd5f9152e8e49d188ff3e16670df51a : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87691fdfc20e4c819ed55fea43603b95 : SUCCESS in 7m 23s\n- networking-baremetal-multitenant-vlans https://zuul.opendev.org/t/openstack/build/d4441b90d44e4726a6a551840a04d68f : TIMED_OUT in 2h 04m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f62a7209725f489fb0755a99fb5a5664 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"c6c03f16497613ed5ac73b0605dea0f521d3861e","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-03-31 12:42:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e826b26c848820f90ee545914440f2d798883b4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 14:51:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a0f125dd8e649dfaa8b53a0215f9a92 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b71c953e9b140c6909e788ce07767b8 : SUCCESS in 7m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1b871fbcf7904e68bbdc5b3227df36e4 : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a612806bd5e84dffbb068f386badc711 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb5d76cb9a4f4d69b4f81e599a35ce33 : SUCCESS in 8m 52s\n- networking-baremetal-multitenant-vlans https://zuul.opendev.org/t/openstack/build/449164ff969a468384d492f6354c8939 : TIMED_OUT in 2h 06m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca6a986f504547809e01f811a2e14d94 : SUCCESS in 7m 49s","accounts_in_message":[],"_revision_number":1},{"id":"d64ffeed960bcc0568ee34c48201c336f2251d28","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-04-02 07:39:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"34e4a0b186d450a037d8be02b6f990c50ccc2bcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 09:48:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfb0432ef6b3445bb9a76cdd9c04411c : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fe260e830e046f7846f0e5447c8cdb7 : SUCCESS in 7m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d76a2744b19e4e9399fcd7ae01898d66 : SUCCESS in 7m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/298d756310f345eb89b6ed0847319e06 : FAILURE in 7m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78f0c3fcc7bc45be859d0536b5d19b13 : SUCCESS in 7m 53s\n- networking-baremetal-multitenant-vlans https://zuul.opendev.org/t/openstack/build/35876e8e8bd540a289c0644e16dac0d3 : TIMED_OUT in 2h 07m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c182f2e701d240d3a53e0a97f43c6f13 : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":2},{"id":"0d72a3a23f03f3b2e3d05cb5f9c09f684b2d6e7e","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-04-03 07:44:45.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"f9d3aab32391b3f138a5d62525df6d7acd0a3a70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 10:08:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88fc3100c6a84956bfb57142fd2d6e70 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ef69dcfef7540a885f1c857b475dec9 : SUCCESS in 8m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/24da708b9b61401c88abdfeb9410b4a5 : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f998cf103f1640d1bc5557285e717514 : FAILURE in 12m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3eaaca088f13468f95e5f5e52ba91a49 : SUCCESS in 7m 40s\n- networking-baremetal-multitenant-vlans https://zuul.opendev.org/t/openstack/build/cf50b75887634ef0b215d0c7b2b348ee : FAILURE in 2h 07m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75d435e15c6d4d1192f57ff91449cd49 : SUCCESS in 7m 28s","accounts_in_message":[],"_revision_number":3},{"id":"9d6b9ec6fc14b33718c377b1484fb3203195513e","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-04-03 10:11:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"11113d5c880d8ef1516ff36fbc789fc77a4e0298","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-04-03 11:37:47.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f79cf51a008e881f915af1f0659cf89d5a5daea7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 13:44:43.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39e3bb72d5b54c189f8c5a1b4a8ec796 : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91fdd96b452b4ebc89730a0393930f31 : SUCCESS in 6m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ec7bbdfddf20405f916c33fb52deba85 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ba56352835443b3930ddd948470817b : FAILURE in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36f79d70546148c98088c3c97b03200e : SUCCESS in 8m 53s\n- networking-baremetal-multitenant-vlans https://zuul.opendev.org/t/openstack/build/bf2fca48e2ed4c0abdd111c98e8150fe : FAILURE in 2h 01m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46d99687df634fac89433fa366b2bed6 : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":5},{"id":"5b7dfac9e6122b620e2490df566835c8cc78a670","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-04-04 14:52:55.000000000","message":"Abandoned\n\nFixed by https://review.opendev.org/717020","accounts_in_message":[],"_revision_number":5},{"id":"1aeedb69456e5ea98fc72886b198e1d3fc638ac8","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-04-09 07:58:56.000000000","message":"Restored\n\nWill use to update the ram","accounts_in_message":[],"_revision_number":5},{"id":"37b3fbd4da9a432fcd342ddac42a3bf3a0302457","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-04-09 08:02:08.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2e4d1cca837982d3f9a67c5843532d589ccd634b","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-09 08:16:28.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"3a59f87b8c8fb75382e59a8c44036fd561ac7c96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 09:59:39.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cc7560a5caa4a10bc8948d22111acc5 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bea57d04f5445f1843c0d999b993faa : SUCCESS in 7m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a64270504ab1489c8b885b9d482740bc : SUCCESS in 8m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d643a529c1f4c26af1193a58e75a1c7 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d39c062fa44408ab7e90ce6bd1c03f2 : SUCCESS in 8m 45s\n- networking-baremetal-multitenant-vlans https://zuul.opendev.org/t/openstack/build/5b594409504f456e83ba36c6f6a71207 : SUCCESS in 1h 35m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33d2fbaa588b43c182eff5fd18944a60 : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":6},{"id":"61f13c95dd8996622704566a3f8a7c47f83c4140","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-09 13:12:36.000000000","message":"Patch Set 6: Code-Review-1\n\nactually, this is moving to dib, so we probably don\u0027t need this","accounts_in_message":[],"_revision_number":6},{"id":"5126ce0274644f62a8af38ae2b081571c7bc83d5","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-04-21 14:24:06.000000000","message":"Abandoned\n\nThe job will be with dib.","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"4bdd37a941f10a50233487520e1edd3e1d5c1226","revisions":{"7fddf93593eb63cb150cb2bb84fd32b1348e767d":{"kind":"REWORK","_number":1,"created":"2020-03-27 12:19:55.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/11/715411/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-baremetal","ref":"refs/changes/11/715411/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/11/715411/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/11/715411/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/11/715411/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-baremetal refs/changes/11/715411/1"}}},"commit":{"parents":[{"commit":"b06ee1ce490120392851c507e2745d92ed6385df","subject":"Enforce running tox with correct python version based on env","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/b06ee1ce490120392851c507e2745d92ed6385df"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-03-27 12:19:35.000000000","tz":60},"committer":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-03-27 12:19:35.000000000","tz":60},"subject":"[DNM] Testing CI","message":"[DNM] Testing CI\n\nChange-Id: I351fbf0a7937ef7b8f86e3653b0e3388e253ff32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/7fddf93593eb63cb150cb2bb84fd32b1348e767d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/7fddf93593eb63cb150cb2bb84fd32b1348e767d"}]},"branch":"refs/heads/master"},"8acc331c9cc3b707bcedf799768f85cc8e79a2b3":{"kind":"REWORK","_number":2,"created":"2020-04-02 07:39:43.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/11/715411/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-baremetal","ref":"refs/changes/11/715411/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/11/715411/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/11/715411/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/11/715411/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-baremetal refs/changes/11/715411/2"}}},"commit":{"parents":[{"commit":"b06ee1ce490120392851c507e2745d92ed6385df","subject":"Enforce running tox with correct python version based on env","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/b06ee1ce490120392851c507e2745d92ed6385df"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-03-27 12:19:35.000000000","tz":60},"committer":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-04-02 07:39:12.000000000","tz":120},"subject":"[DNM] Testing CI","message":"[DNM] Testing CI\n\nChange-Id: I351fbf0a7937ef7b8f86e3653b0e3388e253ff32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/8acc331c9cc3b707bcedf799768f85cc8e79a2b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/8acc331c9cc3b707bcedf799768f85cc8e79a2b3"}]},"branch":"refs/heads/master"},"712c5297e97d7d5d54755bd23df6123e5709fdad":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-04-03 07:44:45.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/11/715411/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-baremetal","ref":"refs/changes/11/715411/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/11/715411/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/11/715411/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/11/715411/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-baremetal refs/changes/11/715411/3"}}},"commit":{"parents":[{"commit":"b06ee1ce490120392851c507e2745d92ed6385df","subject":"Enforce running tox with correct python version based on env","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/b06ee1ce490120392851c507e2745d92ed6385df"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-03-27 12:19:35.000000000","tz":60},"committer":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-04-03 07:43:10.000000000","tz":120},"subject":"[DNM] Testing CI","message":"[DNM] Testing CI\n\nDepends-On: https://review.opendev.org/#/c/717020/\nChange-Id: I351fbf0a7937ef7b8f86e3653b0e3388e253ff32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/712c5297e97d7d5d54755bd23df6123e5709fdad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/712c5297e97d7d5d54755bd23df6123e5709fdad"}]},"branch":"refs/heads/master"},"d97499e9a778f2c3934b51027346209dc9523ede":{"kind":"REWORK","_number":4,"created":"2020-04-03 10:11:12.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/11/715411/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-baremetal","ref":"refs/changes/11/715411/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/11/715411/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/11/715411/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/11/715411/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-baremetal refs/changes/11/715411/4"}}},"commit":{"parents":[{"commit":"b06ee1ce490120392851c507e2745d92ed6385df","subject":"Enforce running tox with correct python version based on env","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/b06ee1ce490120392851c507e2745d92ed6385df"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-03-27 12:19:35.000000000","tz":60},"committer":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-04-03 09:48:45.000000000","tz":120},"subject":"[DNM] Testing CI","message":"[DNM] Testing CI\n\nDepends-On: https://review.opendev.org/#/c/717020/\nChange-Id: I351fbf0a7937ef7b8f86e3653b0e3388e253ff32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/d97499e9a778f2c3934b51027346209dc9523ede"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/d97499e9a778f2c3934b51027346209dc9523ede"}]},"branch":"refs/heads/master"},"9628400add615e41ec0142d63026a6eb2e079c3f":{"kind":"REWORK","_number":5,"created":"2020-04-03 11:37:47.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/11/715411/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-baremetal","ref":"refs/changes/11/715411/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/11/715411/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/11/715411/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/11/715411/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-baremetal refs/changes/11/715411/5"}}},"commit":{"parents":[{"commit":"b06ee1ce490120392851c507e2745d92ed6385df","subject":"Enforce running tox with correct python version based on env","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/b06ee1ce490120392851c507e2745d92ed6385df"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-03-27 12:19:35.000000000","tz":60},"committer":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-04-03 11:37:24.000000000","tz":120},"subject":"[DNM] Testing CI","message":"[DNM] Testing CI\n\nDepends-On: https://review.opendev.org/#/c/717020/\nChange-Id: I351fbf0a7937ef7b8f86e3653b0e3388e253ff32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/9628400add615e41ec0142d63026a6eb2e079c3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/9628400add615e41ec0142d63026a6eb2e079c3f"}]},"branch":"refs/heads/master"},"4bdd37a941f10a50233487520e1edd3e1d5c1226":{"kind":"REWORK","_number":6,"created":"2020-04-09 08:02:08.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/11/715411/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-baremetal","ref":"refs/changes/11/715411/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/11/715411/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/11/715411/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/11/715411/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-baremetal refs/changes/11/715411/6"}}},"commit":{"parents":[{"commit":"12d7aab08bcfa96d1f215f6e8577ce83dc181d4e","subject":"Merge \"Use mock from unittest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/12d7aab08bcfa96d1f215f6e8577ce83dc181d4e"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-03-27 12:19:35.000000000","tz":60},"committer":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-04-09 08:00:52.000000000","tz":120},"subject":"Set ram value for tinyipa job","message":"Set ram value for tinyipa job\n\nThe default in ironic will change 2048 and tinyipa only requires\n384 ram.\n\nChange-Id: I351fbf0a7937ef7b8f86e3653b0e3388e253ff32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/4bdd37a941f10a50233487520e1edd3e1d5c1226"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/4bdd37a941f10a50233487520e1edd3e1d5c1226"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
