)]}'
{"id":"openstack%2Fnetworking-baremetal~718697","triplet_id":"openstack%2Fnetworking-baremetal~master~I38d4d94160d167af311b47fb2c4f4cc29010680e","project":"openstack/networking-baremetal","branch":"master","topic":"ironic-convert-dib","hashtags":[],"change_id":"I38d4d94160d167af311b47fb2c4f4cc29010680e","subject":"Convert networking-baremetal job to dib","status":"MERGED","created":"2020-04-09 13:14:00.000000000","updated":"2020-05-22 10:45:52.000000000","submitted":"2020-05-22 10:43:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718697-1590144181353-fdce670a","meta_rev_id":"4738a7f729cef9eb41a30f39d1b008abe77e29e0","_number":718697,"virtual_id_number":718697,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-22 10:43:01.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2020-05-21 23:25:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":2,"date":"2020-05-22 08:21:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":1,"date":"2020-05-22 08:21:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-23 13:52:48.000000000","updated_by":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"reviewer":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"state":"REVIEWER"},{"updated":"2020-05-07 09:54:51.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-05-19 00:45:39.000000000","updated_by":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"reviewer":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"state":"REVIEWER"},{"updated":"2020-05-21 23:25:56.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2020-05-22 08:21:04.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2020-05-22 10:43:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d785cf95ced3db25772fe5900da95e79d3d48d0c","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-09 13:14:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"974323119701b19c8e41cc7a02b106610e6993c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 15:22:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbdbc7a8180048fdb4f00e44bf996967 : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e2257694b85407cbf3367ee8aced15b : SUCCESS in 6m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a1669dc32fef4389b55681b6e90f5545 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d785e7ca01684cfc843937f2e9dd326b : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d563cd02e554f7b8581df742f446443 : SUCCESS in 7m 10s\n- networking-baremetal-multitenant-vlans https://zuul.opendev.org/t/openstack/build/00ddef6a8670494a8ff836a5e53737ca : FAILURE in 1h 57m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6be1214d97bd453481f06bc01e451c2c : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":1},{"id":"64e5afacba0e64b2139891da1ea9b557e76ba535","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-09 15:32:40.000000000","message":"Patch Set 1:\n\nrecheck this job actually spins up multiple vms so we might think about leaving it with tinyipa","accounts_in_message":[],"_revision_number":1},{"id":"ebde0866ca9ef14a1bf69b8204e721b9e6892a02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 17:23:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac0e13ebf3f647d7addda92e56f1c75b : SUCCESS in 8m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18748ca8357141e2b80d1274c4a95cbd : SUCCESS in 9m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e90f61392e0e425d8390ac9f4e14acf5 : SUCCESS in 9m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a864355309e48529cbf07ff680c9e44 : SUCCESS in 9m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dcf4fd7bfa041bfbb1271751cf09ef5 : SUCCESS in 7m 56s\n- networking-baremetal-multitenant-vlans https://zuul.opendev.org/t/openstack/build/ac086d64e8e54f2890cae0cde28c80c4 : SUCCESS in 1h 45m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3775304197b4b6e857429b13b47d1e4 : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":1},{"id":"8230cda2513fe312932b031ee4e0a2294c0e4c4e","author":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"date":"2020-04-23 13:52:48.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":1},{"id":"ac50f39b8b36fc8b1f373440149c5400dd96ebe1","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-04-24 12:34:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4537b719ef4d232c3285bd67f55e34f7e9b4136a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 12:34:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b3018e137265feb102b8ac30e625ff726ef40874","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 14:29:29.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/877de2f432d54f7ca27509d666928388 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/052c59b49f9945899abf2d12ba3b5866 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6d9807ac4d54e81b55f5402f3cb147a : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd1261e6a6ce4755803e5b3e91e44380 : SUCCESS in 7m 34s\n- networking-baremetal-multitenant-vlans https://zuul.opendev.org/t/openstack/build/66deaa96903d45d9955abe47ca751082 : FAILURE in 1h 53m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2de82eb054834b1d8a5c5494ee0ace8e : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"c0fe94de5bfd1dd80330332b4625847b829da241","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-27 06:36:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3bf2ec76cc06d1088975174d6fdf7d2d42e11ca8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 08:26:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/938a37bf2c6c419eaebce6b1273e5759 : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb2e36d6f5f14ba188e36d14e960b52e : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2cc5b872e0b14ad6a5feae79a5002829 : SUCCESS in 9m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d35a63d91e2b4cb88ba89bd23474193e : SUCCESS in 9m 37s\n- networking-baremetal-multitenant-vlans https://zuul.opendev.org/t/openstack/build/79ca0273ad4c4aa1bb25d0567ce7a2e8 : SUCCESS in 1h 48m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a827b5788694885af423c9332959cf2 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":1},{"id":"3522167460e987de7bd42beef0913f235fb37903","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 08:26:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cc23ddee2ef5a4b5a78106124150440757600b48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 10:34:06.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a19461d3d78847589b7a63c1a95ceec3 : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6bd207d74fe84348b8a774fd92f711f7 : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a62f412e2e347519ba1aa4800def8c2 : SUCCESS in 7m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c96480440f15482781925ec47e64b140 : SUCCESS in 6m 41s\n- networking-baremetal-multitenant-vlans https://zuul.opendev.org/t/openstack/build/1d32321a09cb41469c8ac355943d45c9 : FAILURE in 1h 48m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a37109e561744645b027d409e4f6ce11 : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":1},{"id":"6a954a2ebb7921ecd31b36420a41f667854a5969","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-28 07:36:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"974e75a45e3d63cc6a995ee936bf9e94d0981747","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 08:56:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0ccaade3b72494d91f34bc32f67b49d : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/404ccca3671b45a28e57d243121e0965 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30613e2577f84d3caedfd2b35f7afd98 : SUCCESS in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9589e5c9f9744d88928373fe63027e8 : SUCCESS in 8m 52s\n- networking-baremetal-multitenant-vlans https://zuul.opendev.org/t/openstack/build/0e87ecec57f742b5907b5e69332d5d2b : EXCEPTION\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46c714e9209b41e68ce3125cf0404ef8 : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":1},{"id":"5b73112f81a6ccb7d11e34056c2a6a958d808a85","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-28 12:16:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3da97fd38d62c6c6f46761f80b3c0d9ca3aec615","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 14:18:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ca510badb304f0d88a92bd52302f0ad : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0eee161b2fc4456842a69cd601f38c8 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8759e44bb7234048947214e70c108892 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c251c7c38cb443a19c1b89f76f1d1f8b : SUCCESS in 7m 46s\n- networking-baremetal-multitenant-vlans https://zuul.opendev.org/t/openstack/build/ea9bb5d5325a4158b38f933735b7c640 : FAILURE in 1h 56m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4022f72e573743d2b20da6117cccd2e7 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"45bbe95ee94496b4fc432a4cfea456307fd31dcf","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-29 07:00:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cf3adbb3133b8affbb9bb37f70f48afa55bfcfd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 08:58:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9b561062a5f4854999d205f85b58ffc : SUCCESS in 8m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15de7db393a3485d81b8164dcac7b0d9 : SUCCESS in 7m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04b6ca4ac1e14a0f8978371e4312c495 : SUCCESS in 9m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4324fec5da0d49c2a59b925f6e596a8a : SUCCESS in 10m 48s\n- networking-baremetal-multitenant-vlans https://zuul.opendev.org/t/openstack/build/a527eeefc9e24139bb124b5b7a3f0b00 : SUCCESS in 1h 54m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a82aeeef6bb4df786db09edb83d9362 : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":2},{"id":"823017afbaa9c9b170911f141d274508c16793fa","author":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"date":"2020-05-05 08:39:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"71fdab7d050f0869cf336a19b0d4d14385f1c450","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-05-07 09:54:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ed0ec99ec363d0cc22a9e5851a419c1e2df27e6d","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-05-11 09:08:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c4e2a1c4a013e958e401f9e74a5b8b6e2412375d","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-05-11 09:09:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"769fdc518d738f32b46f020b02e328eac01480a4","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-05-11 09:10:12.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"98b9a65c1fbdeefde59816d3226fbb189250d6e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 11:20:34.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a06e617cb4a94f0892575d6e605f726a : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3520a709e8d54fcc891f7b5afde70088 : SUCCESS in 9m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/715676b9b60642fda7a1666831e5bc5a : SUCCESS in 10m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5d4a6bf8cf5437cbe5c76d27c5a6fe8 : SUCCESS in 9m 31s\n- networking-baremetal-multitenant-vlans https://zuul.opendev.org/t/openstack/build/50fac013eb704e7887126910ad1f1e64 : FAILURE in 2h 03m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc64f64a670042cbb79f4c62cc0f5e63 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":4},{"id":"43b1ca13067d55b2fc07d6c9acb11392f79ccdce","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-05-11 12:23:36.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"65a20219a56d9a765f7010882a21b06125375d67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 14:29:01.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe853c85ee8942c1bdc74808d3c6d620 : SUCCESS in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbe3303fcd0c426a815086c1d76097d0 : SUCCESS in 8m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7c53bcdacdd429090d8523f9d33a6e3 : SUCCESS in 8m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74f535100b704576892ca04e2034b731 : SUCCESS in 8m 47s\n- networking-baremetal-multitenant-vlans https://zuul.opendev.org/t/openstack/build/56ab817af3394d199edb2ecb56a7ff63 : FAILURE in 1h 58m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbcae5b1a7d14286818cf05607b1e241 : SUCCESS in 8m 42s","accounts_in_message":[],"_revision_number":5},{"id":"790f00b659440685c9f693531c1c4ffb73a7409f","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-05-11 14:46:05.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"cd7f22e4f47ecf9982a1e86c53ccdaf80be4d28a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 17:03:16.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84cf0c1fbb2241c68f87ee27c2f3c948 : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc7eebdca7d54847905c992ea17ed78f : SUCCESS in 7m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b694f51a4bb7489fa20cd3faa65c7ebc : SUCCESS in 7m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe47dd8b73ac45aebb2fbdec4324da22 : SUCCESS in 11m 26s\n- networking-baremetal-multitenant-vlans https://zuul.opendev.org/t/openstack/build/63de4d338759440198746a1a310fd392 : FAILURE in 2h 08m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74357c9f0f944891a6a31e17cf95abfd : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":6},{"id":"7e1788caa23a14d3d0739e6055156d1d571e45c2","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-05-15 17:12:26.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"05e00e465211b8bfa1863488723735dcfb3ff340","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-05-18 13:02:47.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"75c44f269d5ca4ddf6ac745f51995ee102a891e4","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-05-18 20:53:07.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"066e3e9bafa367f35b9c79e8c95be8c61acb533b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 22:49:30.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0272980cd9e4b05bc080f80a672a88c : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7934ef92f4140debd182c48d659582c : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ddf76d71d4c040988699282a2e112da5 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c4f339421fb4c69a8a80bff5db84fad : SUCCESS in 7m 32s\n- networking-baremetal-multitenant-vlans https://zuul.opendev.org/t/openstack/build/7bf3aba223064328bbd89a1de4e37bfe : SUCCESS in 1h 54m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/741a2c39155146ecb0a56bc226df0f54 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":7},{"id":"6198937c40c60545f65f2c99c22f3a633cb0b636","author":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"date":"2020-05-19 00:45:39.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"e345ba48f9bc9a4f2bd30d0ebec2aeb5369c41eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 00:45:54.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"2663bf9b17bccd37f2f827f926c39d07b26ec408","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 02:35:18.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a673d1ae183048a19c0c5b4cf982df5e : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5da050c9e54842e89a8aa4f541662b17 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1498ec87212f4693af05529925d64e05 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd7950c6f4534f38930fb0ea02657904 : SUCCESS in 7m 42s\n- networking-baremetal-multitenant-vlans https://zuul.opendev.org/t/openstack/build/d085b44e1a4641dfaa95ba731c5adf39 : FAILURE in 1h 48m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1cb7b9072ab74eb0aa1658ca41804912 : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":7},{"id":"29652ab9760cbf9b695382f21d17e7630dafa523","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-05-19 08:20:10.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"8f50c6127246d2341bfd54b422f4f433894095c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 10:23:59.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9dae5f2ae5648e68f83e568cbd696b3 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddfaa8e0d69c44f7b0d5537f7b90f55f : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf144aa479ac48b7b9ca0b7bf5c94b25 : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5af2f05979d4e4286ec9f75b7e89c15 : SUCCESS in 7m 51s\n- networking-baremetal-multitenant-vlans https://zuul.opendev.org/t/openstack/build/6516015894244bde8e0de39ae7f02c3c : FAILURE in 1h 57m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ebb5ae49b74b43bc837343c47b7649dc : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":7},{"id":"a0f01d61854ca737a736f1a1fbb66e7e62686df4","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-05-19 14:15:39.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"814c869a4d6c2ac9fe729c2d3d93aeda5f2f01be","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-05-19 14:16:29.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"97216aed0b0476a39baf34f2cd207eeca2767f4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 16:42:42.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b939dddcb0d43e3adcceb4a007a2bd2 : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11ac9c2b2a984cb194bf1149843ce827 : SUCCESS in 8m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a04137a62fa94a04857b16daeb955e45 : SUCCESS in 10m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ae2bab354434c4691ad29a8058c0374 : SUCCESS in 15m 05s\n- networking-baremetal-multitenant-vlans https://zuul.opendev.org/t/openstack/build/8546e0ffb25f4d2299c9fb731cc9dabe : SUCCESS in 2h 22m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/906bcc9c983b464cbb60aed49dc6ae98 : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":9},{"id":"7ffaa6c37bab59ca722f1d6451a3add89e1f050e","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-05-21 23:25:56.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"ba5437916941d341573b512023890ee77d77e1c5","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-05-22 08:21:04.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"fb69b241a1f388afd27508f6bf8269e170d5e3be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 08:21:18.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"dd06d1579e8a9575bf01becf37e838d46c8a4dfc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 10:43:01.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d4dccd800da44629a7ea4719abc9394 : SUCCESS in 9m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8454d106adb4edcbdd45d4d0151e005 : SUCCESS in 11m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77080bd9676e4c9dba08f818ce4aec2e : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b21d2f78d7f841a5a388c600a2efa325 : SUCCESS in 10m 12s\n- networking-baremetal-multitenant-vlans https://zuul.opendev.org/t/openstack/build/0ba5d308b08f40edb0d9773e129550d8 : SUCCESS in 2h 15m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a468294937a846c69d5253ab2f268403 : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":9},{"id":"b983c543f08942d02b5042a6508edbdf31871c08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 10:43:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"50220c2e98386d54e620690d955da61190d10ba9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 10:45:52.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e298fa4242ed47198d19aceb83194a59 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"873511fdd407c92bc49bc940fcbc3d941d43e49d","revisions":{"49fcae3f0eb5973658cbb93bf7135612797d970a":{"kind":"REWORK","_number":1,"created":"2020-04-09 13:14:00.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/97/718697/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-baremetal","ref":"refs/changes/97/718697/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/1"}}},"commit":{"parents":[{"commit":"12d7aab08bcfa96d1f215f6e8577ce83dc181d4e","subject":"Merge \"Use mock from unittest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/12d7aab08bcfa96d1f215f6e8577ce83dc181d4e"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-04-09 13:13:11.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-04-09 13:13:11.000000000","tz":120},"subject":"Convert networking-baremetal job to dib","message":"Convert networking-baremetal job to dib\n\nDepends-On: https://review.opendev.org/718607\nChange-Id: I38d4d94160d167af311b47fb2c4f4cc29010680e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/49fcae3f0eb5973658cbb93bf7135612797d970a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/49fcae3f0eb5973658cbb93bf7135612797d970a"}]},"branch":"refs/heads/master"},"5602a33b60699706cb24bf76ec2e718ad39ceff5":{"kind":"REWORK","_number":2,"created":"2020-04-29 07:00:08.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/97/718697/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-baremetal","ref":"refs/changes/97/718697/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/2"}}},"commit":{"parents":[{"commit":"2c63eaa3fea1191d6c565159e26a8b19bc3497a2","subject":"Add py38 package metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/2c63eaa3fea1191d6c565159e26a8b19bc3497a2"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-04-09 13:13:11.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-04-29 07:00:00.000000000","tz":120},"subject":"Convert networking-baremetal job to dib","message":"Convert networking-baremetal job to dib\n\nDepends-On: https://review.opendev.org/718607\nChange-Id: I38d4d94160d167af311b47fb2c4f4cc29010680e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/5602a33b60699706cb24bf76ec2e718ad39ceff5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/5602a33b60699706cb24bf76ec2e718ad39ceff5"}]},"branch":"refs/heads/master"},"7467179554c2ecc3dac14de1a06e0def06225f85":{"kind":"REWORK","_number":3,"created":"2020-05-11 09:09:35.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/97/718697/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-baremetal","ref":"refs/changes/97/718697/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/3"}}},"commit":{"parents":[{"commit":"2c63eaa3fea1191d6c565159e26a8b19bc3497a2","subject":"Add py38 package metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/2c63eaa3fea1191d6c565159e26a8b19bc3497a2"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-04-09 13:13:11.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-05-11 09:09:24.000000000","tz":120},"subject":"Convert networking-baremetal job to dib","message":"Convert networking-baremetal job to dib\n\nChange-Id: I38d4d94160d167af311b47fb2c4f4cc29010680e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/7467179554c2ecc3dac14de1a06e0def06225f85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/7467179554c2ecc3dac14de1a06e0def06225f85"}]},"branch":"refs/heads/master"},"4b98321b5a7175a7c81a58337980269c236cee03":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-05-11 09:10:12.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/97/718697/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-baremetal","ref":"refs/changes/97/718697/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/4"}}},"commit":{"parents":[{"commit":"7e050db9cc2ac80f0f288ce0aaea45fe09261e2e","subject":"Merge \"BUILD_TIMEOUT is not needed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/7e050db9cc2ac80f0f288ce0aaea45fe09261e2e"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-04-09 13:13:11.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-05-11 09:10:05.000000000","tz":120},"subject":"Convert networking-baremetal job to dib","message":"Convert networking-baremetal job to dib\n\nChange-Id: I38d4d94160d167af311b47fb2c4f4cc29010680e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/4b98321b5a7175a7c81a58337980269c236cee03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/4b98321b5a7175a7c81a58337980269c236cee03"}]},"branch":"refs/heads/master"},"b984337e1ca129a0bfe7f81f2b03766a2e769002":{"kind":"REWORK","_number":5,"created":"2020-05-11 12:23:36.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/97/718697/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-baremetal","ref":"refs/changes/97/718697/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/5"}}},"commit":{"parents":[{"commit":"7e050db9cc2ac80f0f288ce0aaea45fe09261e2e","subject":"Merge \"BUILD_TIMEOUT is not needed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/7e050db9cc2ac80f0f288ce0aaea45fe09261e2e"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-04-09 13:13:11.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-05-11 12:21:57.000000000","tz":120},"subject":"Convert networking-baremetal job to dib","message":"Convert networking-baremetal job to dib\n\nReintroducing BUILD_TIMEOUT to increase waiting time for other\nservices since the job is failing on server create that is\nrelated to nova.\n\nChange-Id: I38d4d94160d167af311b47fb2c4f4cc29010680e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/b984337e1ca129a0bfe7f81f2b03766a2e769002"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/b984337e1ca129a0bfe7f81f2b03766a2e769002"}]},"branch":"refs/heads/master"},"fea4b7437659a4cf4092eb934e287c42cf25816f":{"kind":"REWORK","_number":6,"created":"2020-05-11 14:46:05.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/97/718697/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-baremetal","ref":"refs/changes/97/718697/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/6"}}},"commit":{"parents":[{"commit":"7e050db9cc2ac80f0f288ce0aaea45fe09261e2e","subject":"Merge \"BUILD_TIMEOUT is not needed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/7e050db9cc2ac80f0f288ce0aaea45fe09261e2e"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-04-09 13:13:11.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-05-11 14:45:58.000000000","tz":120},"subject":"Convert networking-baremetal job to dib","message":"Convert networking-baremetal job to dib\n\nReintroducing BUILD_TIMEOUT to increase waiting time for other\nservices since the job is failing on server create that is\nrelated to nova.\n\nChange-Id: I38d4d94160d167af311b47fb2c4f4cc29010680e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/fea4b7437659a4cf4092eb934e287c42cf25816f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/fea4b7437659a4cf4092eb934e287c42cf25816f"}]},"branch":"refs/heads/master"},"ec0f9f005e16dcd65bf8fa27b42078b555c0c919":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-05-15 17:12:26.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/97/718697/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-baremetal","ref":"refs/changes/97/718697/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/7"}}},"commit":{"parents":[{"commit":"87cf218c5c3375773084e64919125565acec48d4","subject":"Merge \"Update lower-constraints.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/87cf218c5c3375773084e64919125565acec48d4"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-04-09 13:13:11.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-05-15 17:12:26.000000000","tz":0},"subject":"Convert networking-baremetal job to dib","message":"Convert networking-baremetal job to dib\n\nReintroducing BUILD_TIMEOUT to increase waiting time for other\nservices since the job is failing on server create that is\nrelated to nova.\n\nChange-Id: I38d4d94160d167af311b47fb2c4f4cc29010680e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/ec0f9f005e16dcd65bf8fa27b42078b555c0c919"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/ec0f9f005e16dcd65bf8fa27b42078b555c0c919"}]},"branch":"refs/heads/master"},"cd770b869f1b5f9fb37056a9daba4078989a6abf":{"kind":"REWORK","_number":8,"created":"2020-05-19 14:15:39.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/97/718697/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-baremetal","ref":"refs/changes/97/718697/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/8"}}},"commit":{"parents":[{"commit":"87cf218c5c3375773084e64919125565acec48d4","subject":"Merge \"Update lower-constraints.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/87cf218c5c3375773084e64919125565acec48d4"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-04-09 13:13:11.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-05-19 14:15:30.000000000","tz":120},"subject":"Convert networking-baremetal job to dib","message":"Convert networking-baremetal job to dib\n\nReintroducing BUILD_TIMEOUT to increase waiting time for other\nservices since the job is failing on server create that is\nrelated to nova.\n\nChange-Id: I38d4d94160d167af311b47fb2c4f4cc29010680e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/cd770b869f1b5f9fb37056a9daba4078989a6abf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/cd770b869f1b5f9fb37056a9daba4078989a6abf"}]},"branch":"refs/heads/master"},"873511fdd407c92bc49bc940fcbc3d941d43e49d":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-05-19 14:16:29.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/97/718697/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-baremetal","ref":"refs/changes/97/718697/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-baremetal refs/changes/97/718697/9"}}},"commit":{"parents":[{"commit":"6b6467b7cd1c029838e18981574da5ecd4e6a720","subject":"Merge \"Bump hacking version to 3.0.0 and fix pep8 test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/6b6467b7cd1c029838e18981574da5ecd4e6a720"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-04-09 13:13:11.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-05-19 14:15:49.000000000","tz":120},"subject":"Convert networking-baremetal job to dib","message":"Convert networking-baremetal job to dib\n\nReintroducing BUILD_TIMEOUT to increase waiting time for other\nservices since the job is failing on server create that is\nrelated to nova.\n\nChange-Id: I38d4d94160d167af311b47fb2c4f4cc29010680e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/873511fdd407c92bc49bc940fcbc3d941d43e49d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/873511fdd407c92bc49bc940fcbc3d941d43e49d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
