)]}'
{"id":"openstack%2Fnetworking-baremetal~856357","triplet_id":"openstack%2Fnetworking-baremetal~master~I827dabeb774cd3bdcd8bf845aee70f3a62d22d0f","project":"openstack/networking-baremetal","branch":"master","topic":"dnm","hashtags":[],"change_id":"I827dabeb774cd3bdcd8bf845aee70f3a62d22d0f","subject":"DNM - Testing CI before Zed release","status":"ABANDONED","created":"2022-09-08 01:33:13.000000000","updated":"2022-11-16 17:40:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8fbc53308e6383b98eaff37bb75d90b5172ca5a2","_number":856357,"virtual_id_number":856357,"owner":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-09-08 03:41:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-08 03:41:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b27000eea7f2cc836446968ae19513cfde08c91d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2022-09-08 01:33:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9332d555e375d8fb252c18941654f530a2c0218","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 03:41:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccbe9ce9f24b43eaad281f21ba672d5c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9d7e8f97304d4b2db425128208e5a0ac : SUCCESS in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3518296670c4a0b9fe1ac908b0450f1 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25719aaa67fc412c8a3f9718be11b2d9 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e1ed07572ec4a11b0147eb44dc12777 : SUCCESS in 6m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf0606d6358a426cab4476a9a228c166 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/071408a8e162460088ad968af63ccd1a : SUCCESS in 9m 55s\n- networking-baremetal-multitenant-vlans https://zuul.opendev.org/t/openstack/build/fe6a20161a564c8c9c327d2f05204c83 : SUCCESS in 2h 02m 49s","accounts_in_message":[],"_revision_number":1},{"id":"8fbc53308e6383b98eaff37bb75d90b5172ca5a2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-11-16 17:40:57.000000000","message":"Abandoned\n\nNot needed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1a8b791e122f390f7690a398f4adcac1a3a1315b","revisions":{"1a8b791e122f390f7690a398f4adcac1a3a1315b":{"kind":"REWORK","_number":1,"created":"2022-09-08 01:33:13.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/57/856357/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-baremetal","ref":"refs/changes/57/856357/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/57/856357/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/57/856357/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/57/856357/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-baremetal refs/changes/57/856357/1"}}},"commit":{"parents":[{"commit":"8c23e7e98540869b45d54831eac1c6dc37aee010","subject":"Merge \"Doc - network device configuration capabilities\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/8c23e7e98540869b45d54831eac1c6dc37aee010"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2022-09-08 01:32:38.000000000","tz":-180},"committer":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2022-09-08 01:33:12.000000000","tz":-180},"subject":"DNM - Testing CI before Zed release","message":"DNM - Testing CI before Zed release\n\nChange-Id: I827dabeb774cd3bdcd8bf845aee70f3a62d22d0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/1a8b791e122f390f7690a398f4adcac1a3a1315b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/1a8b791e122f390f7690a398f4adcac1a3a1315b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
