)]}'
{"id":"openstack%2Fnetworking-baremetal~932736","triplet_id":"openstack%2Fnetworking-baremetal~master~I73982059cca0411fb1ed6c64f03031b8dcd668d4","project":"openstack/networking-baremetal","branch":"master","attention_set":{},"removed_from_attention_set":{"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2024-10-21 14:24:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I73982059cca0411fb1ed6c64f03031b8dcd668d4","subject":"add pyproject.toml to support pip 23.1","status":"MERGED","created":"2024-10-18 17:17:34.000000000","updated":"2024-10-21 14:25:27.000000000","submitted":"2024-10-21 14:24:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"932736","meta_rev_id":"3793a754343f2237e2b2a9dd719189d1718f78ae","_number":932736,"virtual_id_number":932736,"owner":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-21 14:24:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2024-10-18 17:18:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-21 12:51:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-21 12:51:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-18 17:18:14.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2024-10-18 19:09:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-21 12:51:48.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"335ec6275a03c3fc7e04cc789835e984ba75becf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-10-18 17:17:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6aaabb5b5f6c3e1d66f80eede9124448a78335d5","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2024-10-18 17:18:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5769b094fcb20d08c9223fbe724a1ad16ce05705","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 19:09:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a39499dc57dc4c6fb3df9be46c875914\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6276a1639cc046a18b545855cf6c0b83 : SUCCESS in 1m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/00a9dc5ec369448584bfe3efb74b74cd : SUCCESS in 1m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e537129051294680bad024ef6cee5990 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3fda2afc9a7428c8ec3f2768960b9d9 : SUCCESS in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/542e9594c9a14c19b0cb09af1bc237b7 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ce2986cefb646c1b809ef0487944679 : SUCCESS in 8m 24s\n- networking-baremetal-multitenant-vlans https://zuul.opendev.org/t/openstack/build/a99149778cf541d2999cf947bd7d31ed : SUCCESS in 1h 42m 18s\n- networking-baremetal-tox-codespell https://zuul.opendev.org/t/openstack/build/e92b6bb5fad24760a7469b9ab8a95d03 : SUCCESS in 2m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b5d4f8ae8664ec271bdb51c093d7d8360f37985c","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2024-10-21 12:51:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7732e9f9fdad764b15d016161d669d8b0b0a07b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 12:52:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"796d818a06e556a55baa6fc85f12746186cd555e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 14:24:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/917bca774a5c4ae9906c9920049816ba\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a56dc0e7072e4f66b0214316fa49bd04 : SUCCESS in 1m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/da8b9a2f8971488f81b4a1c9afa73dc4 : SUCCESS in 1m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a12c8442a944a6b889b8c4535deb3ca : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e12851cdbdd4767be80ed7c632acae8 : SUCCESS in 4m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aabfc8967c994874b30db60fc7ee3dcd : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90676cd04a76421ca1f2212de4743132 : SUCCESS in 7m 44s\n- networking-baremetal-multitenant-vlans https://zuul.opendev.org/t/openstack/build/d7f07d3153ea482f965702b9ae3be544 : SUCCESS in 1h 32m 04s","accounts_in_message":[],"_revision_number":1},{"id":"a33d813a2c30b71643e2192168b760254ab89f12","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 14:24:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3793a754343f2237e2b2a9dd719189d1718f78ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 14:25:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69f669a7000c484d8d01546cd087c840\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d71a6982d0ce485cb08a5b0145dccea5 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1d62d4e96109576e1f9681980f93e810d1750881","revisions":{"1d62d4e96109576e1f9681980f93e810d1750881":{"kind":"REWORK","_number":1,"created":"2024-10-18 17:17:34.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/36/932736/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-baremetal","ref":"refs/changes/36/932736/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/36/932736/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/36/932736/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-baremetal refs/changes/36/932736/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-baremetal refs/changes/36/932736/1"}}},"commit":{"parents":[{"commit":"185e6c105cfb553be07fc8dc81f5c2ea33969c02","subject":"Merge \"avoid attribute error on bad password or config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/185e6c105cfb553be07fc8dc81f5c2ea33969c02"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2024-10-18 17:17:22.000000000","tz":-300},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2024-10-18 17:17:22.000000000","tz":-300},"subject":"add pyproject.toml to support pip 23.1","message":"add pyproject.toml to support pip 23.1\n\npip 23.1 removed the \"setup.py install\" for projects that do not have\ntheir own pyproject.toml by having a hardcoded one within pip.\n\nTo address that, this change adds the minimal pyproject.toml\nto enable pbr to be properly used to build editable wheels.\n\nSee https://pip.pypa.io/en/stable/news/#v23-1\nand https://github.com/pypa/pip/issues/8368 for more info.\n\nChange-Id: I73982059cca0411fb1ed6c64f03031b8dcd668d4\nSigned-off-by: Doug Goldstein \u003ccardoe@cardoe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/1d62d4e96109576e1f9681980f93e810d1750881"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-baremetal/commit/1d62d4e96109576e1f9681980f93e810d1750881"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
