)]}'
{"id":"openstack%2Fnetworking-bgpvpn~1004445","triplet_id":"openstack%2Fnetworking-bgpvpn~master~I8ef2d0d0b8a416c1913b691d88bc91dc6cb39d14","project":"openstack/networking-bgpvpn","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-10 15:28:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8ef2d0d0b8a416c1913b691d88bc91dc6cb39d14","subject":"Remove Python 3.10 support","status":"MERGED","created":"2026-09-07 15:46:00.000000000","updated":"2026-09-10 15:30:16.000000000","submitted":"2026-09-10 15:28:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004445","meta_rev_id":"5fa49cd635cb97b2f7ca07b31f97975965796082","_number":1004445,"virtual_id_number":1004445,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-10 15:28:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-09-08 06:13:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-09-10 13:45:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-09-10 13:45:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-08 06:13:53.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-09-08 12:27:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-10 13:45:23.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"8fbae4a39b32390f49dd0e61b5d7f197db9756e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-07 15:46:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1dc52bd06dff4f1c75fecdbe30061507ea9fc08f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-08 06:13:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"43de88c73c5f34e6ef9eafe29754e349a710fdba","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-08 11:38:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4ddfd561bb5eee4cc5876cd08d3af6850d5cced0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 12:27:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf73c8e60dfe4e42a3a08c144b2a0610\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c718dbf3b4914bb0a0c316aff7b7723b : SUCCESS in 1m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/093fcc1fc19d4e23a646fb482f853143 : SUCCESS in 9m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/766e0bfe0f054af989d16cd7cebd6baf : SUCCESS in 7m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b76d256ab1fb44f5a9d7243dbe2bc0ce : SUCCESS in 4m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fefadfbed81846a4839ea0d690bacf97 : SUCCESS in 10m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46485c1ef760479e8a750dbed1f1e7f1 : SUCCESS in 9m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6965418fa7845b2a5ff74057a981877 : SUCCESS in 5m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6a1f8676bfc5406888a711177add1f95 : SUCCESS in 3m 16s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/251b115836b34934985058756dadef64 : SUCCESS in 7m 50s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/45ece17602c247aeadca6c86bec92ce7 : SUCCESS in 7m 58s\n- horizon-tox-python3-django60 https://zuul.opendev.org/t/openstack/build/f7e512bd10144075a6f684b51080b4d5 : FAILURE in 7m 25s (non-voting)\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/d672b51690724f6cac5cd7ced08ebb84 : SUCCESS in 48m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c035b055894c49428753ac165dd0ec84 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"851f90b4d9bae94d1d8e2d8eb4afcd27a12cbcab","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-09-10 13:45:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"46b04b5821045413e109a6d2020a40240e46971a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 13:46:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6ed6fe2c13fc60ba68af64f47eb144933b5bf316","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 15:28:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb7ffc4085d34a4687c178436aa5edde\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dd4187f3069b4c4c983189408317d8bb : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9fd5feecf4f47e486f6a6478bee0401 : SUCCESS in 7m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cafecaf73daa4203a7dd521697e3751a : SUCCESS in 8m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e6deb8d759584e7ea5943b7011ca9d63 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3033a5458fe46888eadaa2718d9da67 : SUCCESS in 9m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6426a4229f8d421f99e619539784e9f6 : SUCCESS in 6m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/20270a9b07674516ac0cb6f887746472 : SUCCESS in 3m 16s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/2dbc945a7c3c450a8a0a19bac02653e8 : SUCCESS in 4m 23s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/d9d4c70479a548d9a3a438fb3af90c39 : SUCCESS in 9m 15s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/3ed6e963c1e547e0a89fbe0db0b7fb85 : SUCCESS in 1h 36m 11s","accounts_in_message":[],"_revision_number":1},{"id":"190856648f233a208d30ba2e36ff702df0560b2f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 15:28:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5fa49cd635cb97b2f7ca07b31f97975965796082","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 15:30:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f3bfb8ec3db42baa7b7d8d164032d01\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d40419ebe8d141bb80bc8dc1bf00b647 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc525210abd44bda6d5b01b12b0226a33234f7d9","revisions":{"fc525210abd44bda6d5b01b12b0226a33234f7d9":{"kind":"REWORK","_number":1,"created":"2026-09-07 15:46:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/1004445/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/45/1004445/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/45/1004445/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/45/1004445/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/45/1004445/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/45/1004445/1"}}},"commit":{"parents":[{"commit":"13c86c38b9ca7efecea7db64570644d63045f5cc","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/13c86c38b9ca7efecea7db64570644d63045f5cc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-07 15:45:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-07 15:45:48.000000000","tz":540},"subject":"Remove Python 3.10 support","message":"Remove Python 3.10 support\n\nPython 3.10 was removed from tested runtimes during 2026.2 cycle.\n\nChange-Id: I8ef2d0d0b8a416c1913b691d88bc91dc6cb39d14\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/fc525210abd44bda6d5b01b12b0226a33234f7d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/fc525210abd44bda6d5b01b12b0226a33234f7d9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
