)]}'
{"id":"openstack%2Fnetworking-bgpvpn~281321","triplet_id":"openstack%2Fnetworking-bgpvpn~master~Ie085e2f2e1c03ccbf0f95405c6f6a53d73e1db38","project":"openstack/networking-bgpvpn","branch":"master","hashtags":[],"change_id":"Ie085e2f2e1c03ccbf0f95405c6f6a53d73e1db38","subject":"Fixing pylint upgrade issues","status":"MERGED","created":"2016-02-17 15:03:36.000000000","updated":"2016-02-17 15:16:33.000000000","submitted":"2016-02-17 15:16:33.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"102955c87b8321953bba58667c277b0d19b599a9","_number":281321,"virtual_id_number":281321,"owner":{"_account_id":13933,"name":"Niko Hermanns","email":"nikolas.hermanns@ericsson.com","username":"enikher"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2016-02-17 15:16:33.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":2888,"name":"Mathieu Rohon","email":"mathieu.rohon@gmail.com","username":"mathieu-rohon"},{"value":0,"_account_id":13933,"name":"Niko Hermanns","email":"nikolas.hermanns@ericsson.com","username":"enikher"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2888,"name":"Mathieu Rohon","email":"mathieu.rohon@gmail.com","username":"mathieu-rohon"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":2,"date":"2016-02-17 15:14:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2888,"name":"Mathieu Rohon","email":"mathieu.rohon@gmail.com","username":"mathieu-rohon"},{"value":1,"date":"2016-02-17 15:15:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13933,"name":"Niko Hermanns","email":"nikolas.hermanns@ericsson.com","username":"enikher"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2888,"name":"Mathieu Rohon","email":"mathieu.rohon@gmail.com","username":"mathieu-rohon"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2016-02-17 15:04:28.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":1,"date":"2016-02-17 15:14:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2888,"name":"Mathieu Rohon","email":"mathieu.rohon@gmail.com","username":"mathieu-rohon"},{"value":0,"_account_id":13933,"name":"Niko Hermanns","email":"nikolas.hermanns@ericsson.com","username":"enikher"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":2888,"name":"Mathieu Rohon","email":"mathieu.rohon@gmail.com","username":"mathieu-rohon"},{"value":0,"_account_id":13933,"name":"Niko Hermanns","email":"nikolas.hermanns@ericsson.com","username":"enikher"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":2888,"name":"Mathieu Rohon","email":"mathieu.rohon@gmail.com","username":"mathieu-rohon"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":13933,"name":"Niko Hermanns","email":"nikolas.hermanns@ericsson.com","username":"enikher"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-02-17 15:04:28.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2016-02-17 15:14:06.000000000","updated_by":{"_account_id":2888,"name":"Mathieu Rohon","email":"mathieu.rohon@gmail.com","username":"mathieu-rohon"},"reviewer":{"_account_id":2888,"name":"Mathieu Rohon","email":"mathieu.rohon@gmail.com","username":"mathieu-rohon"},"state":"REVIEWER"},{"updated":"2016-02-17 15:16:33.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"9781470ba1d75ad1b3e1f2c2576d4c41c599d481","author":{"_account_id":13933,"name":"Niko Hermanns","email":"nikolas.hermanns@ericsson.com","username":"enikher"},"date":"2016-02-17 15:03:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3bd6c368b4c40df90262a2c0b6602a32539592d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-17 15:12:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-bgpvpn-pep8 http://logs.openstack.org/21/281321/1/check/gate-networking-bgpvpn-pep8/b08f4f3/ : SUCCESS in 2m 11s\n- gate-networking-bgpvpn-docs http://docs-draft.openstack.org/21/281321/1/check/gate-networking-bgpvpn-docs/5158980//doc/build/html/ : SUCCESS in 1m 56s\n- gate-networking-bgpvpn-python27 http://logs.openstack.org/21/281321/1/check/gate-networking-bgpvpn-python27/d54c6d2/ : SUCCESS in 2m 11s","accounts_in_message":[],"_revision_number":1},{"id":"aaf34fb85f73de0513d54702e83bb382ccad2a84","author":{"_account_id":2888,"name":"Mathieu Rohon","email":"mathieu.rohon@gmail.com","username":"mathieu-rohon"},"date":"2016-02-17 15:14:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4377e73008497a54d831adac77630f0f4f60e0d5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-17 15:14:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"19520243bab7401de6cd28c60d916b92c200e34d","author":{"_account_id":13933,"name":"Niko Hermanns","email":"nikolas.hermanns@ericsson.com","username":"enikher"},"date":"2016-02-17 15:15:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bcb938b04ebaa327492823fa13a57c3ec4d0bd54","date":"2016-02-17 15:16:33.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"102955c87b8321953bba58667c277b0d19b599a9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-17 15:16:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-networking-bgpvpn-docs http://docs-draft.openstack.org/21/281321/1/gate/gate-networking-bgpvpn-docs/895cb4a//doc/build/html/ : SUCCESS in 1m 35s\n- gate-networking-bgpvpn-pep8 http://logs.openstack.org/21/281321/1/gate/gate-networking-bgpvpn-pep8/8cd8313/ : SUCCESS in 1m 59s\n- gate-networking-bgpvpn-python27 http://logs.openstack.org/21/281321/1/gate/gate-networking-bgpvpn-python27/edbddd4/ : SUCCESS in 2m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"401a0437922895086e5bfd28f630b78130cae1d6","revisions":{"401a0437922895086e5bfd28f630b78130cae1d6":{"kind":"REWORK","_number":1,"created":"2016-02-17 15:03:36.000000000","uploader":{"_account_id":13933,"name":"Niko Hermanns","email":"nikolas.hermanns@ericsson.com","username":"enikher"},"ref":"refs/changes/21/281321/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/21/281321/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/21/281321/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/21/281321/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/21/281321/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/21/281321/1"}}},"commit":{"parents":[{"commit":"e21d03eb217726504d6e07b778b6d52afb06af73","subject":"bagpipe: agent, update RPC setup code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/e21d03eb217726504d6e07b778b6d52afb06af73"}]}],"author":{"name":"Nikolas Hermanns","email":"nikolas.hermanns@ericsson.com","date":"2016-02-17 14:58:10.000000000","tz":60},"committer":{"name":"Nikolas Hermanns","email":"nikolas.hermanns@ericsson.com","date":"2016-02-17 14:59:30.000000000","tz":60},"subject":"Fixing pylint upgrade issues","message":"Fixing pylint upgrade issues\n\nChange-Id: Ie085e2f2e1c03ccbf0f95405c6f6a53d73e1db38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/401a0437922895086e5bfd28f630b78130cae1d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/401a0437922895086e5bfd28f630b78130cae1d6"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e21d03eb217726504d6e07b778b6d52afb06af73","is_merged_in_target_branch":true,"change_id":"I19c9bb83dc2519cdc2b0d8e0e11858e80d36cd4a","change_number":275213,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
