)]}'
{"id":"openstack%2Fnetworking-bgpvpn~604335","triplet_id":"openstack%2Fnetworking-bgpvpn~master~I0e15124db206dafde51166e33e3284e86252d936","project":"openstack/networking-bgpvpn","branch":"master","topic":"horizon-extras","hashtags":[],"change_id":"I0e15124db206dafde51166e33e3284e86252d936","subject":"Use extras for horizon dependency","status":"MERGED","created":"2018-09-21 12:00:52.000000000","updated":"2020-02-26 23:09:46.000000000","submitted":"2020-02-24 10:32:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"604335-1582540346296-eaa21e4c","meta_rev_id":"88df58ec44e8350a5dc60d0a20821cbed300c889","_number":604335,"virtual_id_number":604335,"owner":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"date":"2020-02-20 21:55:39.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-24 10:32:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"all":[{"value":2,"date":"2020-02-24 08:58:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2020-02-21 13:47:09.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2020-02-24 10:05:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-20 21:55:39.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-02-24 08:58:49.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2020-02-24 10:05:44.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-02-24 10:32:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e799cf159aa7b1689a70408927e2562e51764588","author":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"date":"2018-09-21 12:00:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c43f8b6a0032aac9ff08d6a0bf3d90ded29d7c7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-21 12:45:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/35/604335/1/check/openstack-tox-pep8/d532d6b/ : SUCCESS in 7m 58s\n- openstack-tox-py27 http://logs.openstack.org/35/604335/1/check/openstack-tox-py27/0ababa4/ : SUCCESS in 8m 17s\n- openstack-tox-py35 http://logs.openstack.org/35/604335/1/check/openstack-tox-py35/0710ae3/ : SUCCESS in 8m 36s\n- build-openstack-sphinx-docs http://logs.openstack.org/35/604335/1/check/build-openstack-sphinx-docs/77ec5dc/html/ : SUCCESS in 5m 07s\n- build-openstack-releasenotes http://logs.openstack.org/35/604335/1/check/build-openstack-releasenotes/3cafd4b/html/ : SUCCESS in 3m 58s\n- requirements-check http://logs.openstack.org/35/604335/1/check/requirements-check/23e1c2d/ : SUCCESS in 3m 11s\n- test-release-openstack-python3 http://logs.openstack.org/35/604335/1/check/test-release-openstack-python3/27799cb/ : SUCCESS in 2m 59s\n- openstack-tox-cover http://logs.openstack.org/35/604335/1/check/openstack-tox-cover/5819ed2/cover/ : SUCCESS in 9m 23s\n- legacy-networking-bgpvpn-dsvm-functional http://logs.openstack.org/35/604335/1/check/legacy-networking-bgpvpn-dsvm-functional/722a39a/ : SUCCESS in 16m 29s\n- openstack-tox-py36 http://logs.openstack.org/35/604335/1/check/openstack-tox-py36/599c41b/ : SUCCESS in 8m 47s\n- openstack-tox-docs http://logs.openstack.org/35/604335/1/check/openstack-tox-docs/9826def/html/ : SUCCESS in 4m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/35/604335/1/check/openstack-tox-lower-constraints/1b8320e/ : SUCCESS in 11m 13s\n- horizon-openstack-tox-python3-django111 http://logs.openstack.org/35/604335/1/check/horizon-openstack-tox-python3-django111/7cc5258/ : SUCCESS in 6m 32s\n- networking-bgpvpn-dsvm-functional http://logs.openstack.org/35/604335/1/check/networking-bgpvpn-dsvm-functional/de1ec1c/ : SUCCESS in 15m 15s","accounts_in_message":[],"_revision_number":1},{"id":"809d2066c2f574b1d3e0e46fe3bb47575c95b9b3","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-05 05:27:09.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nI am fine with this change. Please fix random spaces.","accounts_in_message":[],"_revision_number":1},{"id":"87a2692d6533d0e7ea9b766c6915bdf699cd628d","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-02-20 21:54:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"06ecee105c13828789908368080e1ee6e3fb11d7","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-02-20 21:54:45.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"615cb42a6abc90a0ec1a773d7df98b09ab3e7282","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-02-20 21:55:39.000000000","message":"Patch Set 2:\n\nAfter investigating the cause of the failure of networking-bagpipe, I started to think it is reasonable and nice to have.","accounts_in_message":[],"_revision_number":2},{"id":"3f7dda4a564887de30db42ed3728fbacf3c81f2a","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-02-20 21:56:40.000000000","message":"Topic set to horizon-extras","accounts_in_message":[],"_revision_number":2},{"id":"a088e8bd74b4e1d6b0adb7f201d766e0bc886b2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 23:48:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/505ddc99e3ab48e1bf591f19049cae87 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d9ed6ca14024bf3aa549b61313de270 : SUCCESS in 9m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/891dcde9ffd54fda81da7f2e4b5dd792 : SUCCESS in 9m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8afcfcfd7d8e48d281a7917ea05663f1 : SUCCESS in 10m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2633effed924fdd9cfd8eb5a7361ad5 : FAILURE in 8m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eecd3fe2d49b43ab9189c2eb87fbc388 : SUCCESS in 8m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ae23eaf92994ab59abc2dd027a0d0b0 : SUCCESS in 6m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6826ff89ecf04688a90c515aded2bd56 : SUCCESS in 4m 28s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/241a1a3b28df4da6a88d61e9db1a1bbb : SUCCESS in 1h 49m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7f22e5600144b94a149d93f8a7c385f : SUCCESS in 10m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62c6afd3e2364a03a914345b050e132f : SUCCESS in 8m 50s\n- networking-bgpvpn-dsvm-functional https://zuul.opendev.org/t/openstack/build/aa9dc92acbff4579a9e3bcf74095994f : SUCCESS in 16m 24s","accounts_in_message":[],"_revision_number":2},{"id":"eeab211a25cd10e1af808d2cb0fbd0525d3e7e98","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-21 13:47:01.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nIt makes life more difficult, but I can live with it.\n\nStrange that py38 fails with using old neutron, and neutron-lib.","accounts_in_message":[],"_revision_number":2},{"id":"433dd694b5dae9f916b9e463040f4f69c3bb0332","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-21 13:47:09.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0a865ec228985ee03fda068db851a17a4f22484d","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-02-24 08:58:49.000000000","message":"Patch Set 2: Code-Review+2\n\nYes, I think it\u0027s highly relevant to switch horizon to \u0027extra\u0027.","accounts_in_message":[],"_revision_number":2},{"id":"db5dab133b46bf1bd71a0e324b4dbd20bfe7bb63","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-02-24 10:05:44.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9ac5798e05c32ee14afe70fdf99adc6ae212dc8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 10:05:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"45e3c3791b5e1eda6a54e751aedc2564face070c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 10:32:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"22ea3cf46ee045dcdab5580339fa0fd9e732b7b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 10:32:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5bea256800a6462db32f4a983202000c : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b70f027360c440a0a6f3e2116bbee079 : SUCCESS in 9m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6281881b2a54d39a4766c8c1b531211 : SUCCESS in 10m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0812b4d6a82744b4813af8a8f9186e2e : SUCCESS in 10m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f9bed1838fa42908fe43682a19fcaf7 : SUCCESS in 7m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a11cbeb337254e4b8d9254dad1697744 : SUCCESS in 5m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/294855b4a66241d88cf25356432e694a : SUCCESS in 6m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9042be0f7ac44054900d9716cd3bde80 : SUCCESS in 11m 43s\n- networking-bgpvpn-dsvm-functional https://zuul.opendev.org/t/openstack/build/9da6fbd95f444825a2b984eb22ddd519 : SUCCESS in 20m 03s","accounts_in_message":[],"_revision_number":2},{"id":"bccc675da7ba69e4b6021972c6f8a86954ea20d0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 10:33:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c308977a2dc4ca190180c881a4e4f46 : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15c3f1ab5f4640de933cb6db18105aa7 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2},{"id":"88df58ec44e8350a5dc60d0a20821cbed300c889","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-02-26 23:09:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"af146b9d337e9a0ad59686e3ea3797ee7d861479","revisions":{"7f8de5650d2bc74c0e9a2aeb9d96216efd042d4d":{"kind":"REWORK","_number":1,"created":"2018-09-21 12:00:52.000000000","uploader":{"_account_id":21883,"name":"Manuel Buil","email":"mbuil@suse.com","username":"mbuil"},"ref":"refs/changes/35/604335/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/35/604335/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/35/604335/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/35/604335/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/35/604335/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/35/604335/1"}}},"commit":{"parents":[{"commit":"f881d7f52cf12d450a9154abb50d771ba76a7da2","subject":"Merge \"Import legacy jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/f881d7f52cf12d450a9154abb50d771ba76a7da2"}]}],"author":{"name":"Manuel Buil","email":"mbuil@suse.com","date":"2018-09-21 12:01:08.000000000","tz":120},"committer":{"name":"Manuel Buil","email":"mbuil@suse.com","date":"2018-09-21 12:02:35.000000000","tz":120},"subject":"Use extras for horizon dependency","message":"Use extras for horizon dependency\n\nHorizon is not a key dependency for networking-bgpvpn and right now it\nforces projects to install horizon if they want to use networking-bgpvpn\nwhich does not make sense. This patch follows the same phylosophy as:\n\nhttps://review.openstack.org/#/c/602622/7\n\nChange-Id: I0e15124db206dafde51166e33e3284e86252d936\nSigned-off-by: Manuel Buil \u003cmbuil@suse.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/7f8de5650d2bc74c0e9a2aeb9d96216efd042d4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/7f8de5650d2bc74c0e9a2aeb9d96216efd042d4d"}]},"branch":"refs/heads/master"},"af146b9d337e9a0ad59686e3ea3797ee7d861479":{"kind":"REWORK","_number":2,"created":"2020-02-20 21:54:40.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/35/604335/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/35/604335/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/35/604335/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/35/604335/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/35/604335/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/35/604335/2"}}},"commit":{"parents":[{"commit":"7b0056927648e1eb51f7103d0e98336e3d578073","subject":"Merge \"Remove references for unittest2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/7b0056927648e1eb51f7103d0e98336e3d578073"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2018-09-21 12:01:08.000000000","tz":120},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-02-20 21:54:21.000000000","tz":540},"subject":"Use extras for horizon dependency","message":"Use extras for horizon dependency\n\nHorizon is not a key dependency for networking-bgpvpn and right now it\nforces projects to install horizon if they want to use networking-bgpvpn\nwhich does not make sense.\n\"extras\" allows us to declare optional dependencies [1].\n\nnetworking-bagpipe is an optional dependency and\nit is specified as extras too.\n\n[1] https://setuptools.readthedocs.io/en/latest/setuptools.html#declaring-extras\n\nChange-Id: I0e15124db206dafde51166e33e3284e86252d936\nCo-Authored-By: Akihiro Motoki \u003camotoki@gmail.com\u003e\nSigned-off-by: Manuel Buil \u003cmbuil@suse.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/af146b9d337e9a0ad59686e3ea3797ee7d861479"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/af146b9d337e9a0ad59686e3ea3797ee7d861479"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
