)]}'
{"id":"openstack%2Fnetworking-bgpvpn~696952","triplet_id":"openstack%2Fnetworking-bgpvpn~master~Id8046ef85da72207097fef52b7966a4b3a067704","project":"openstack/networking-bgpvpn","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"Id8046ef85da72207097fef52b7966a4b3a067704","subject":"[ussuri][goal] Drop python 2.7 support and testing","status":"MERGED","created":"2019-12-02 20:12:43.000000000","updated":"2019-12-05 06:26:48.000000000","submitted":"2019-12-05 06:24:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696952-1575527044780-232123cc","meta_rev_id":"35e4a6d3faec557eb8c26d5a5bc9b35970a488a8","_number":696952,"virtual_id_number":696952,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-12-04 13:23:56.000000000","_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-05 06:24:04.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2019-12-04 13:25:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2019-12-05 04:53:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2019-12-05 04:53:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-03 05:46:01.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-12-04 13:25:15.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2019-12-05 04:53:40.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-12-05 06:24:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"66bcb8e1bd105503e8984053a57302c4053b13fe","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-12-02 20:12:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6815b2ac7cb64c10df53723faeeb537eb32c349b","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-12-02 20:14:24.000000000","message":"Topic set to drop-py27-support","accounts_in_message":[],"_revision_number":1},{"id":"42c3dfc1132efae586d4e374c8efed8ed3564ead","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 22:01:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a68408ab49494a3097dfd8bb328e9c78 : SUCCESS in 2m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0372e9b6aeef4b2eb54caa75836445c6 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/211ce98e137243c4a32b56ca38c5afc8 : SUCCESS in 7m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0d02d6daea034116ab26f262d71bc775 : SUCCESS in 7m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6fff7434c2b416bb265007977ca32f0 : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea91fb7e9d85485ea5dfdc1336af1a43 : SUCCESS in 3m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3b3390633387450f8ad933dfff384427 : SUCCESS in 3m 12s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/6ad2cb9a765b4d4180e804f5891b3ade : SUCCESS in 8m 02s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/73ec09d8ffbb46b3ab2c8060cbd80d97 : SUCCESS in 7m 30s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/6d8868fa410e466ba1c3561f26c30c58 : SUCCESS in 1h 47m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e118224761bb47bb99839d54842043bf : SUCCESS in 11m 49s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/a5ccf0b961b548b9b09a1241c0e52d9e : SUCCESS in 1h 26m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbb3ed6b90544f9aae2d96d34453bf80 : SUCCESS in 11m 30s\n- networking-bgpvpn-dsvm-functional https://zuul.opendev.org/t/openstack/build/1290d4c08d7f4d35824225674ab36a3c : SUCCESS in 12m 58s","accounts_in_message":[],"_revision_number":1},{"id":"b3229ed5dc5fa062d7211025a968f7c778a70893","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-03 07:39:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"26827b8f2cbe40e0d3d0418d9be9bab862ef5c09","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-12-03 07:44:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fdeac74bba5fa9bb44c58c5bc3ba82769a2b4ab7","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-12-03 10:16:24.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0c583abd5ba3f327dc09c1b47b47ca9b64e648a2","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-12-03 11:09:38.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"cf447c6880de1d0c75fc9ed7e82828996722431b","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-12-03 11:10:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0a0b34bb0a645e17bfbfb6bbae1fa8610de2d38e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-12-03 12:28:12.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks for taking care for this","accounts_in_message":[],"_revision_number":2},{"id":"ab753dc5f98490d18783ccc20cc6c3676a6559d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 13:26:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1e62e0bf01254dc69bddbd7b36cd305a : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e766e8da9113413f8c1bb7d24a47ab14 : SUCCESS in 7m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb1e2eda418f4e75a63ca675eef6a033 : SUCCESS in 8m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a0717825d4ef40949d0d591d58983ffd : SUCCESS in 8m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/751519ebfef9465dbf2d4c2302e8e6ff : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b4bbf3a0d894907ac7a51969b635a4b : SUCCESS in 4m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/eddae77b4cb14e788ef7ce00766b73ea : SUCCESS in 3m 00s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/60b4b5b6a0c6474ab237ca5748815855 : SUCCESS in 6m 47s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/da0ef9af22de4aba95b16fa660164db6 : SUCCESS in 8m 03s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/876dd197fa1c4018bafed42996c44e8c : SUCCESS in 2h 14m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8467c840cf647c9859c45288b1bac26 : SUCCESS in 11m 30s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/c920fc12edb14829b65c835c33b1a369 : SUCCESS in 1h 22m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bcd392f0820444788896f437a6b409e4 : SUCCESS in 9m 21s\n- networking-bgpvpn-dsvm-functional https://zuul.opendev.org/t/openstack/build/03e96d97dc4648089667ff88a3416b1e : SUCCESS in 21m 31s","accounts_in_message":[],"_revision_number":2},{"id":"60c54d30f33e486f30847369eb60cccf0c4ed2f9","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-04 02:52:23.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9055c3854a53ac18252bf105157c007c10ed55ef","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-12-04 06:28:17.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"777ad0da9ba086d76dd13d3cffba5987f432195e","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-12-04 06:29:10.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8350857cb5c35f2343f27787ad0bf53c122f526e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 08:46:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/567c66455b934dd0909ba8cc7a5872e6 : SUCCESS in 2m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25f2eb982b0a4cdd84af6d530503ecc0 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/137ffa457f9644cfa2ccc3e840851491 : SUCCESS in 8m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/478e88f0f9ca4bdc8c3ba7c35ed0d5ce : SUCCESS in 8m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/172bbc2918414805a3d63bdfb653a580 : SUCCESS in 7m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1fb68bd4cf7422cb57ee0fdccaed07b : SUCCESS in 5m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8936afb5698e4e639875271703566bd2 : SUCCESS in 5m 15s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/dc0f4d2afe63466e9af2651d3c6a68ca : SUCCESS in 7m 06s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/5d35b572bba54c37b896d434906232fa : SUCCESS in 8m 50s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/91d428eb30074beeac7f638c80434af6 : SUCCESS in 2h 15m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b3d864c522e4dc1a870e6ce6b2b8fe9 : SUCCESS in 14m 34s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/401fb6422dce4dc0b41aeb008663c3d4 : SUCCESS in 1h 23m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61c0289c0e774a608fd52cf97bbe1754 : SUCCESS in 10m 13s\n- networking-bgpvpn-dsvm-functional https://zuul.opendev.org/t/openstack/build/0f07e0a9505f44dfa5a7b9e954cd0798 : SUCCESS in 36m 25s","accounts_in_message":[],"_revision_number":3},{"id":"494e04f26ddbbd25d2145820969b977f20ce015c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-12-04 13:21:10.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nsmall nit regarding whitespace in commit msg","accounts_in_message":[],"_revision_number":3},{"id":"da517a9cd3da89acbff6e37b844c83ebd5fba78a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-12-04 13:21:40.000000000","message":"Patch Set 3:\n\nVishal Manchanda \u0026 Akihiro: thanks for working on this","accounts_in_message":[],"_revision_number":3},{"id":"02f1301ab5feaf51796c8c580001acdb37e071e3","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-12-04 13:23:27.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"114997ac1b9b87a0c315a90c1c2e2b9d14a65575","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-12-04 13:23:56.000000000","message":"Patch Set 4:\n\n\u003e Vishal Manchanda \u0026 Akihiro: thanks for working on this\n\nwelcome.","accounts_in_message":[],"_revision_number":4},{"id":"1dd927483a2cbcaf20d4137641e529e80beb0e18","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-12-04 13:24:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e5e12b4f245aaaf0a2d7772947951d84c36793c3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-12-04 13:25:15.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"7e571951e3d70bca37b2f7a500f07f39d398a140","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 16:15:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e7baf7aa3edc49a1a4e590af8f490efa : SUCCESS in 2m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/692ed4756e24420a8ba9aab9b06c93ea : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc01bc3ee5dd47548677fd58b574d3d8 : SUCCESS in 7m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/59a269286b834d58ba4e186ab0e3b1c2 : SUCCESS in 9m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec4d11b80f3e4a2da5f1c8dfe7486ea3 : SUCCESS in 7m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46f8089f883b4cb69587bc2c730a34bd : SUCCESS in 5m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f6ae359df386442d8b9835c367b533dc : SUCCESS in 3m 03s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/22f71b7fa60642938de6b52df3b65525 : SUCCESS in 6m 17s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/112a1db742cb4a158f308a2555c2065a : SUCCESS in 6m 58s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/f3fbbe2ebd4d4b60864aac16de9a55a8 : FAILURE in 2h 47m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21a44fe32c474a5faf4a736a292dc68e : SUCCESS in 10m 55s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/1e5aefeebfe443218eab2ba0ac556931 : SUCCESS in 1h 35m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd37a46834ea41be9e95914ec0b9f5da : SUCCESS in 7m 21s\n- networking-bgpvpn-dsvm-functional https://zuul.opendev.org/t/openstack/build/b5374658f25f40d1bddf71174032d9d3 : SUCCESS in 14m 23s","accounts_in_message":[],"_revision_number":4},{"id":"5dac112927113b801741c80928efdb990164ad97","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-05 04:53:40.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"a7a9a80559f07529dcaf130edbddf1dcd5c8f1ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 04:53:55.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f7f894025a57b23c951d24d3e6978d76be361fb1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 06:24:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"64ea18f58d8716d4a325b69b6d4f5f7dfc4ab2a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 06:24:04.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8a4b74bfa5f741498f39883365fc9a6c : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdaede597a854ab6a6c30210ebcc4ba9 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e82c29b58464152b9e704999b346eed : SUCCESS in 6m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/85f726bd74a24d61a4da2ea3be8b5e30 : SUCCESS in 8m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e0a1dc2b8e840b1b2b6d29d44b7a3ae : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/686cc69fa37a4fccb766080de1b7e1a7 : SUCCESS in 3m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0077f570944c4c69b191d90b1f131170 : SUCCESS in 3m 28s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/0f9609f700b8440293a659b84a3fa1d7 : SUCCESS in 6m 32s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/13dd4211d75f4619b6d99641870e8c6f : SUCCESS in 8m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8bc9ceb59e84efcb449c0616dc1ce46 : SUCCESS in 9m 42s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/9f933794ab9b41179a9e98c06c66b3aa : SUCCESS in 1h 28m 17s\n- networking-bgpvpn-dsvm-functional https://zuul.opendev.org/t/openstack/build/2dda8cc4d25f49328549f39112e50ea8 : SUCCESS in 16m 33s","accounts_in_message":[],"_revision_number":4},{"id":"35e4a6d3faec557eb8c26d5a5bc9b35970a488a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 06:26:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8e1077412884c82b82437e87a6ca5fe : SUCCESS in 1m 25s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/121cf0418fb74b52aadeacc8fea1997c : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"40fa37be8c80f9e59d32c561e441a367907e8f85","revisions":{"cc2e387b34d521637ae1f10157dfde4f88611cc7":{"kind":"REWORK","_number":1,"created":"2019-12-02 20:12:43.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/52/696952/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/52/696952/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/52/696952/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/52/696952/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/52/696952/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/52/696952/1"}}},"commit":{"parents":[{"commit":"6c905092817ff372393989ea50169eb4746c9a5a","subject":"Merge \"use standard_attr db from neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/6c905092817ff372393989ea50169eb4746c9a5a"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2019-12-02 20:10:50.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2019-12-02 20:10:50.000000000","tz":0},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nnetworking-bgpvpn is ready with python 3 and ok to drop the\npython 2.7 support.\n\nAlso this patch update minimum tox version in tox.ini file.\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nChange-Id: Id8046ef85da72207097fef52b7966a4b3a067704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/cc2e387b34d521637ae1f10157dfde4f88611cc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/cc2e387b34d521637ae1f10157dfde4f88611cc7"}]},"branch":"refs/heads/master"},"a8c28ce8da4b237dc38eb2a0636912e3b97ea305":{"kind":"REWORK","_number":2,"created":"2019-12-03 11:09:38.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/52/696952/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/52/696952/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/52/696952/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/52/696952/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/52/696952/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/52/696952/2"}}},"commit":{"parents":[{"commit":"6c905092817ff372393989ea50169eb4746c9a5a","subject":"Merge \"use standard_attr db from neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/6c905092817ff372393989ea50169eb4746c9a5a"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2019-12-02 20:10:50.000000000","tz":0},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2019-12-03 11:09:32.000000000","tz":0},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nnetworking-bgpvpn is ready with python 3 and ok to drop the\npython 2.7 support.\n\nAlso this patch update minimum tox version in tox.ini file.\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nChange-Id: Id8046ef85da72207097fef52b7966a4b3a067704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/a8c28ce8da4b237dc38eb2a0636912e3b97ea305"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/a8c28ce8da4b237dc38eb2a0636912e3b97ea305"}]},"branch":"refs/heads/master"},"ebbe98309ec98a2b16b08c0049b0b6052390bf06":{"kind":"REWORK","_number":3,"created":"2019-12-04 06:28:17.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/52/696952/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/52/696952/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/52/696952/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/52/696952/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/52/696952/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/52/696952/3"}}},"commit":{"parents":[{"commit":"6c905092817ff372393989ea50169eb4746c9a5a","subject":"Merge \"use standard_attr db from neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/6c905092817ff372393989ea50169eb4746c9a5a"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2019-12-02 20:10:50.000000000","tz":0},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2019-12-04 06:28:04.000000000","tz":0},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nnetworking-bgpvpn is ready with python 3 and ok to drop the\npython 2.7 support.\n\nAlso this patch update minimum tox version and remove \u0027pypy\u0027 entry \nas it is no longer run in tox.ini file.\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nChange-Id: Id8046ef85da72207097fef52b7966a4b3a067704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/ebbe98309ec98a2b16b08c0049b0b6052390bf06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/ebbe98309ec98a2b16b08c0049b0b6052390bf06"}]},"branch":"refs/heads/master"},"40fa37be8c80f9e59d32c561e441a367907e8f85":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-12-04 13:23:27.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/52/696952/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/52/696952/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/52/696952/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/52/696952/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/52/696952/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/52/696952/4"}}},"commit":{"parents":[{"commit":"6c905092817ff372393989ea50169eb4746c9a5a","subject":"Merge \"use standard_attr db from neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/6c905092817ff372393989ea50169eb4746c9a5a"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2019-12-02 20:10:50.000000000","tz":0},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2019-12-04 13:22:42.000000000","tz":0},"subject":"[ussuri][goal] Drop python 2.7 support and testing","message":"[ussuri][goal] Drop python 2.7 support and testing\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nnetworking-bgpvpn is ready with python 3 and ok to drop the\npython 2.7 support.\n\nAlso this patch update minimum tox version and remove \u0027pypy\u0027 entry\nas it is no longer run in tox.ini file.\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nChange-Id: Id8046ef85da72207097fef52b7966a4b3a067704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/40fa37be8c80f9e59d32c561e441a367907e8f85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/40fa37be8c80f9e59d32c561e441a367907e8f85"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
