)]}'
{"id":"openstack%2Fnetworking-bgpvpn~720951","triplet_id":"openstack%2Fnetworking-bgpvpn~master~I2cf3afe0ce833e5d85563a6b10f9ac71f3b8c485","project":"openstack/networking-bgpvpn","branch":"master","topic":"unittest.mock","hashtags":[],"change_id":"I2cf3afe0ce833e5d85563a6b10f9ac71f3b8c485","subject":"Use unittest.mock instead of third party mock","status":"MERGED","created":"2020-04-18 16:55:43.000000000","updated":"2020-05-25 09:00:08.000000000","submitted":"2020-05-25 08:58:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720951-1590397084659-da4d1bf7","meta_rev_id":"b7cff2477b198ed8865d7bf5b5745a52b99c6021","_number":720951,"virtual_id_number":720951,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-25 08:58:04.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"all":[{"value":2,"date":"2020-05-25 07:29:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":2,"date":"2020-05-20 11:00:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"all":[{"value":1,"date":"2020-05-25 07:29:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-20 11:00:57.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-05-25 07:29:55.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2020-05-25 08:58:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d5e4244bf85113f4b37b95210c3ee40becdc74bb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-18 16:55:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d53eb45dcaa4c22337863caf8bcbe87525996c12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-18 19:00:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73f37824f1ea4c9aaa8a8e7521701e7e : SUCCESS in 10m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0562d2e428a45689bcf2220da40b96d : FAILURE in 10m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/806964cfa456472fac332c2269b2783c : FAILURE in 11m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68b43b4cf1654e099de67beaa486cbf2 : FAILURE in 12m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf71ab01798347eb93bec59145ff6dd1 : SUCCESS in 9m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4ae20fbcb9c400690c310be7b1fd403 : FAILURE in 14m 10s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/aefeeec0ea2d488eb3caed5528611452 : SUCCESS in 1h 29m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3650fe63ab7443ddb54418270b20d4bf : SUCCESS in 11m 14s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/6d8978fe46d44560a45c269c51949290 : FAILURE in 19m 16s","accounts_in_message":[],"_revision_number":1},{"id":"8c0c9cdd380e77c18b6b6118e8daa887436119cb","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-19 06:59:34.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d0f96141b5baa4c683a696059a1e9307922bebb3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-19 07:28:42.000000000","message":"Patch Set 2: Code-Review-1\n\nCould you please remove mock from test-requirements.txt and lower-constraints.txt ?","accounts_in_message":[],"_revision_number":2},{"id":"3829bc10b940cc07cdacf6ec48616acab58167de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 08:28:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe9eeefb2b9b44529db02fe4f4d03a05 : SUCCESS in 8m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82a50ad4454341d88dec1dee00e49aa6 : SUCCESS in 8m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01e18e144dd14112868f652c52dcfb27 : SUCCESS in 9m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83a8880b8a6848df99adcb859ce43411 : SUCCESS in 10m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49b170ca55a8485d88aba4a9c33fef6d : SUCCESS in 15m 28s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/4be41ffc7eb74edf8d62dcc1720d9079 : SUCCESS in 1h 24m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aef602be5c4b42e8a73e14c4708dea93 : SUCCESS in 8m 24s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/55f629caaca14d51ba94b82e363fe74e : SUCCESS in 21m 53s","accounts_in_message":[],"_revision_number":2},{"id":"48488e84ebc021669a4afc9a4461e2a0cbf566b6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-19 11:23:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8d0959505668540b46f56dd6c65c3a137b4de513","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 12:54:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66f7121c6d24482bbbb8b8b6a3d0643d : SUCCESS in 9m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07df703046f340a08c0bf58ac5711ef8 : SUCCESS in 8m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c923805f44d4792930a4a3501cdf654 : SUCCESS in 10m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc4b7e82d1374718aed0d9afa064117f : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfed6e8a806d40c68c66047d7ff1eed6 : SUCCESS in 3m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3080a29ece354bc9ac16da0ce61ea54e : SUCCESS in 3m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33e438767d624e068d3598d2d304c855 : SUCCESS in 11m 26s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/9b1a0169d3e242aa945c9ffcbeedee92 : SUCCESS in 1h 25m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f918d6f144a475c9de77f679d095f2e : SUCCESS in 8m 44s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/14805a1658a143009669706e36676872 : SUCCESS in 20m 34s","accounts_in_message":[],"_revision_number":3},{"id":"f07ff7299e1b7ddade4f0f399a3a3d355ccf0dc6","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-20 11:00:57.000000000","message":"Patch Set 3: Code-Review+2\n\nThank you very much","accounts_in_message":[],"_revision_number":3},{"id":"1fcc9c9f9ca1499331519496e049e0b6c0e6b789","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-05-25 07:29:55.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6940fd6675ea2afbaac96065c21642ad5b452c24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 07:30:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"20d59972466df9b319ec7044ed511f761e05d3af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 08:58:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e0e3fc67101af3e7befafdb15f48289738d2a157","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 08:58:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5019b0d7fcd645e6bb381c216e4b644e : SUCCESS in 8m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0fc3196af514fa5951c53ab2e3523c2 : SUCCESS in 10m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51afa5a321234ffdaadc223ebf3c70b0 : SUCCESS in 8m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/952753f49f0b47b99da8318378a0fcfb : SUCCESS in 7m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1cb462b690134ea9a10f9320e468f865 : SUCCESS in 3m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7b6158564da6450581e9435e6893dc9e : SUCCESS in 3m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81bb6eb77d2543aea3f2b9e9a52b3bad : SUCCESS in 10m 27s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/340eafc5819f4f2885c107b5fe2faf58 : SUCCESS in 1h 22m 52s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/fef49c35b23947c0a59d7599d6b84cba : SUCCESS in 17m 02s","accounts_in_message":[],"_revision_number":3},{"id":"799b6e01a787fd584ece43b0617d6bce8883fb8f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 09:00:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f3767b6726a4b0ab538215f85a11721 : SUCCESS in 1m 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/934c8b0cf99c4df9b40661b21703e898 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e15e405e2f80c81101dfa56c0014dc4ba5c504fe","revisions":{"319942480c3c2dbabcdf57ce6c1b17c9398c1138":{"kind":"REWORK","_number":1,"created":"2020-04-18 16:55:43.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/51/720951/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/51/720951/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/51/720951/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/51/720951/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/51/720951/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/51/720951/1"}}},"commit":{"parents":[{"commit":"d6a93f62337ffd4db88337b627008c6d9a7d276e","subject":"Fix recent gate failures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/d6a93f62337ffd4db88337b627008c6d9a7d276e"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-18 16:55:32.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-18 16:55:32.000000000","tz":-300},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib.\n\nChange-Id: I2cf3afe0ce833e5d85563a6b10f9ac71f3b8c485\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/319942480c3c2dbabcdf57ce6c1b17c9398c1138"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/319942480c3c2dbabcdf57ce6c1b17c9398c1138"}]},"branch":"refs/heads/master"},"df1920bf3c6b69968c65578e77f0f53d347daa73":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-19 06:59:34.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/51/720951/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/51/720951/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/51/720951/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/51/720951/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/51/720951/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/51/720951/2"}}},"commit":{"parents":[{"commit":"17527ef261cf48a8c8471d83728a0e51bc3de5c8","subject":"Merge \"mock out ProviderConfiguration.__init__\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/17527ef261cf48a8c8471d83728a0e51bc3de5c8"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-18 16:55:32.000000000","tz":-300},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-05-19 06:59:34.000000000","tz":0},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib.\n\nChange-Id: I2cf3afe0ce833e5d85563a6b10f9ac71f3b8c485\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/df1920bf3c6b69968c65578e77f0f53d347daa73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/df1920bf3c6b69968c65578e77f0f53d347daa73"}]},"branch":"refs/heads/master"},"e15e405e2f80c81101dfa56c0014dc4ba5c504fe":{"kind":"REWORK","_number":3,"created":"2020-05-19 11:23:52.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/51/720951/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/51/720951/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/51/720951/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/51/720951/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/51/720951/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/51/720951/3"}}},"commit":{"parents":[{"commit":"17527ef261cf48a8c8471d83728a0e51bc3de5c8","subject":"Merge \"mock out ProviderConfiguration.__init__\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/17527ef261cf48a8c8471d83728a0e51bc3de5c8"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-18 16:55:32.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-19 11:23:45.000000000","tz":-300},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib.\n\nChange-Id: I2cf3afe0ce833e5d85563a6b10f9ac71f3b8c485\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/e15e405e2f80c81101dfa56c0014dc4ba5c504fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/e15e405e2f80c81101dfa56c0014dc4ba5c504fe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
