)]}'
{"id":"openstack%2Fnetworking-bgpvpn~722040","triplet_id":"openstack%2Fnetworking-bgpvpn~master~I653f3d863bcdbeac6874a6e51e1e86ea16a18c78","project":"openstack/networking-bgpvpn","branch":"master","topic":"add-victoria-python-jobtemplates","hashtags":[],"change_id":"I653f3d863bcdbeac6874a6e51e1e86ea16a18c78","subject":"Add Python3 victoria unit tests","status":"MERGED","created":"2020-04-22 16:06:28.000000000","updated":"2020-05-05 08:48:50.000000000","submitted":"2020-05-05 08:44:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722040-1588668281080-56a7dbf5","meta_rev_id":"ab470de7e70f2276dc78116597e884b1746f4590","_number":722040,"virtual_id_number":722040,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-05 08:44:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"recommended":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2020-05-04 12:53:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-05-04 12:56:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-05-05 07:22:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-05-05 07:22:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-22 16:45:31.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-04-23 15:22:39.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-05-04 12:53:38.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-05-04 12:56:37.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2020-05-05 07:22:03.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-05-05 08:44:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"68d59ef23aae2ad81a4877684a2c781395da6e8b","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-22 16:06:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cedba3418e6a290abe17dfec9d4606433d9d5823","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-22 16:45:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b91ce30e0facaf84c2cf050af1643683b687c0be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 17:58:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75b01999175c4781b98d365d0b332d43 : SUCCESS in 11m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15b6268e7ecc4e26b0d8b68c6061fc65 : SUCCESS in 12m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1b040bd0b9b4015b2dcb92acb54e8a2 : SUCCESS in 12m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b034748fab3c4b75bf6d7cda39088f1f : SUCCESS in 10m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8b5eea0ecb86485e943921e28bbd7145 : SUCCESS in 9m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11aad19f6cfe4e4292c5231957d98603 : SUCCESS in 13m 03s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/c6cbf30c08f14b658dee7d1391befa5a : SUCCESS in 1h 26m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14dbdcf18c7d4237936a904ae72b9f2d : SUCCESS in 11m 18s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/55759d109b0d4c7da6c8e0d4ae5c46a7 : FAILURE in 19m 28s","accounts_in_message":[],"_revision_number":1},{"id":"8191dbaa28560bf29baeed3b20afd049c1da6605","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-23 15:22:39.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bb13454176930611df78d7904366679f5ea9eaa3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 17:07:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33ff9dc99d0c4844a58119eadea2a8db : SUCCESS in 8m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b55139aa5c9a46fc956223af33ed47a3 : SUCCESS in 8m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97b40a56872a48a0aaaf47fefaeecabf : SUCCESS in 9m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa368b7703ab48a4b1be259cef16990f : SUCCESS in 6m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7226b0cd549141aaaf6c0bb8987fc859 : SUCCESS in 8m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc91c75655ae4350a1835f56387844be : SUCCESS in 10m 15s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/895a314b1e3444bc931e894184993f5f : SUCCESS in 1h 28m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/740c00ead39c4ddd950f39c2994a97de : SUCCESS in 8m 45s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/2768ece6e5bb408685b4d78459c836e7 : FAILURE in 13m 42s","accounts_in_message":[],"_revision_number":1},{"id":"e27a4e5ad406283a3ac12bfc3f039a7a198cc6d8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-24 09:24:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1db98040e3fcb7ad26a2f4b0a8a807908258c28d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 10:38:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb57f43956c947ab9052e5e466536869 : SUCCESS in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/313e14da3fe04479b2fbad061e5abfe0 : SUCCESS in 7m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25e45c2fa63b4b72b47318b20e7a020e : SUCCESS in 10m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ee7aa7b51c0435db0e221dc29e37c04 : SUCCESS in 7m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ff429e8518341d3bb7c7a8cda2d8b1e : SUCCESS in 10m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29d28983ef1340cfb07fc439e0f0f012 : SUCCESS in 10m 32s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/cc598594adce4cf6835a128f7b5584e3 : SUCCESS in 1h 12m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1a80ac4cd2544b299ad7e006cb530d9 : SUCCESS in 9m 47s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/75da7dea9f9f46e7b9a37552e514a85c : FAILURE in 14m 27s","accounts_in_message":[],"_revision_number":1},{"id":"971ba1ab0f7a5c3a9257589bfd0a294d82f4f428","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-24 13:02:59.000000000","message":"Patch Set 1:\n\nhmmm, functional seems to fail constantly with:\n2020-04-24 09:38:04.797990 | controller |   File \"/usr/local/lib/python3.6/dist-packages/neutron/tests/base.py\", line 29, in \u003cmodule\u003e\n2020-04-24 09:38:04.798000 | controller |     import mock\n2020-04-24 09:38:04.798010 | controller | ModuleNotFoundError: No module named \u0027mock\u0027","accounts_in_message":[],"_revision_number":1},{"id":"c9ab63f02fef2c7fbbb2a6e9bb2f497d77e2fa68","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-24 14:05:57.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"b5c4b7a2daeea11cc4fb0467a442195bdf00eb50","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-24 14:09:45.000000000","message":"Patch Set 2:\n\nrecheck\nLet\u0027s see if adding mock to test-requirements helps....","accounts_in_message":[],"_revision_number":2},{"id":"b0c356bba939299c82cba8b97f4cc26abc09a1f1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-24 14:37:46.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"cf187ceafbb146804d88b06e669cbb575badd367","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 16:21:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d76ac142b8454612b266e45182b18c8e : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db74f7c78c1945518483307d57016c8a : SUCCESS in 11m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3ad55186c054cafada2659dbb4c7661 : SUCCESS in 11m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dc3c24324474e0b83b5e1b3d4f4b633 : SUCCESS in 9m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4ccde08794de42fea873bf89d42b1a06 : SUCCESS in 9m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b8aab2a1fd440c0bdfafc93eef67d0e : SUCCESS in 12m 54s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/8e071be905bd44fa97a4a16313ba4f58 : SUCCESS in 1h 33m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91c229ac318d48f4bdb1671532118121 : SUCCESS in 9m 57s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/001c6380eba94a5cbe659343ccc8d146 : FAILURE in 18m 53s","accounts_in_message":[],"_revision_number":3},{"id":"4305605c6d0335d771193fefe8e1d2d162ae1b3b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-25 05:08:09.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7cd8de4f345870b397b3fbe08b9671550db9faa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 06:41:42.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7225a46183d4249bdd87d07457ca06f : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a089b1e2dfe41418ee9110063e4f12d : SUCCESS in 7m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0dbd92893624de88887cd3460015c71 : SUCCESS in 9m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b17d40d6ac34d47820cf261f0473c49 : SUCCESS in 5m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe29c9e4849642698a72005ea5d678d1 : SUCCESS in 7m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/465f5d24c8d94cbdbb4f1a0fbe3ba484 : SUCCESS in 10m 02s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/7b727136c07741428b56676a88a8a249 : SUCCESS in 1h 28m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe861ee5db3e46dfa58c739b846a0347 : SUCCESS in 8m 12s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/b7d8b5eb61da47c2aefa4445807d2225 : FAILURE in 14m 02s","accounts_in_message":[],"_revision_number":3},{"id":"0415f64469969468a28a79ed8cd6b13a5743d65d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-30 08:17:32.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"4b7f48da7e7fb819326af306b8233115157a96ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 09:48:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e778491916d4a26b4ccd95da915b8a3 : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9154d26a8d8f490d9f9770a01b5f4f1b : SUCCESS in 9m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c1951e25754439fb8da6ee01bfd33e9 : SUCCESS in 8m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c08d12bf4944514be3891cba03de2f2 : SUCCESS in 6m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c37da340fae54c05b9bf74899ce91802 : SUCCESS in 8m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08b64bf5e9fe4321a2bed304aba932de : SUCCESS in 10m 13s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/ab956c63a4614bc6b2d847b175e2717f : SUCCESS in 1h 28m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f236fb31f1574b758e605923dbe21b54 : SUCCESS in 8m 59s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/e58114ca09bd42f896e8b16d9a22a5f0 : SUCCESS in 16m 41s","accounts_in_message":[],"_revision_number":4},{"id":"e1083fe05066295284ae84ee89224def1f658986","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-04-30 10:38:22.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"a320991ca404527236d95995e5ba196d4b027e20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 10:38:39.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1a40a35c9a189625e3d8953fdd4a9a5b3588fca0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 12:09:26.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23eb63fc6c554e99a5f4c110839e95bd : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01847c2432594f2ab1d8de9376c210c0 : SUCCESS in 7m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/977c25fca474462cb46ddf507d9b0038 : FAILURE in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cb9cc5174e049e89149f88b544dfb40 : SUCCESS in 5m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/35fd4977237a437281a3bd48bf749eae : SUCCESS in 8m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8faa26760739443ea1d475f373cd1389 : SUCCESS in 10m 48s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/44a51abbebc84f958ddf8fe55fab5a2e : SUCCESS in 1h 25m 57s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/8eeadb8b849e4e00a232532fb634cb1b : SUCCESS in 18m 40s","accounts_in_message":[],"_revision_number":4},{"id":"067b3764885adc5c6b3eba6ae03da37617e869c5","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-04-30 13:43:41.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"6b81f62e549fbcd5d8133c1b7011a65339338474","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 15:32:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d31a7fb3eb664c3484c389913a53f27e : SUCCESS in 7m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9a96710ed96418b829ca18f418c3dd6 : SUCCESS in 7m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7de0697d8e734336ae10870c6ed189dd : SUCCESS in 8m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a4813dd8b6e445c9273636d99c66159 : SUCCESS in 6m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/69ae82b4d30a46ceb46763181907820f : SUCCESS in 8m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/05af89312deb4cbbbd8594229f41ad0d : SUCCESS in 12m 20s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/5fa444e7740740f6bf328af8aee6991b : SUCCESS in 1h 43m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32d088a8f1964042867d868758c2e5ce : SUCCESS in 8m 25s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/3070b1c9eef44613b5ec995af8ba0a3f : SUCCESS in 19m 22s","accounts_in_message":[],"_revision_number":4},{"id":"0ea3f09a5088a597197eb7bf4cfa34bc51e0f776","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 15:32:54.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"cf9a44e74d562e9f09ce291afc1760e747bfeb90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 17:08:36.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be34b1a7fd9f4caf855912a864d4e59a : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3c546a2250947ed84fed2aa630d7ab5 : SUCCESS in 9m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f602066d44a640e0991aca9d0d64eb41 : FAILURE in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c277133e5ce4e109f5340658c7a14ae : SUCCESS in 7m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0746289714cd494482da7617c0e95ed7 : SUCCESS in 7m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a557fbd792954a1899dddc39337269dd : SUCCESS in 10m 54s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/46601e4d99d14212a481fe4e81897163 : SUCCESS in 1h 28m 55s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/799234dad73a4e849a99184dd067bdd0 : SUCCESS in 23m 56s","accounts_in_message":[],"_revision_number":4},{"id":"53ba23695bcdf2b0ae813f35208354599963765b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-05-01 04:42:25.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"61ab9d6d8d1238c9ba2542ad0368ed04925fee6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 06:15:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ced01438ea3c44ebbfe22a947265a333 : SUCCESS in 8m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9f0ecc872874824aa7a7e70c0e158d7 : SUCCESS in 7m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9281aa0debde4ccb91356e1296db556e : SUCCESS in 9m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f900a588d8142c5b0275334ca797764 : SUCCESS in 6m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/158f4fc99c174e3fafd82877e065c6fd : SUCCESS in 6m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6403c8d9ef84d7d9cb9fddc24f8af25 : SUCCESS in 12m 19s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/d851a2dc58324a65abdf7ce7244b741e : SUCCESS in 1h 31m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13e21b960f644a0ab0c9ef8c12f1cb90 : SUCCESS in 10m 44s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/74e1d77e6ab64521ae38e982848c13fa : SUCCESS in 29m 17s","accounts_in_message":[],"_revision_number":4},{"id":"4881c0163abe3b8747f6141a5bfa4d38b9b96bac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 06:16:13.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0c98c2ab41e51455a687fef52583311e4900144a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 07:39:07.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fbda759032a49e587c70cf30ddf22f0 : SUCCESS in 8m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7684df16bb004d1eb0c64137462f0313 : SUCCESS in 7m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad0e91f6340b46418aaf41d374585662 : FAILURE in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44046229124a4c6987c607a33100f2e1 : SUCCESS in 10m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b9b3ea06c3444b9690cf69a5d15fd608 : SUCCESS in 9m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/88b5784bce9b4da2a8adaed47944efbd : SUCCESS in 9m 54s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/2d033e749e1a41c7b70f2f14e7e358ea : SUCCESS in 1h 21m 35s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/b909f8721b494e03a329f7424a74b508 : SUCCESS in 21m 26s","accounts_in_message":[],"_revision_number":4},{"id":"09517241e479296b764435035b37afcd7b76974e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-02 19:06:31.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"15a1c00aa007988c70a02c89aa85373e439d8426","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 20:30:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/277154d128fb4bac803255599a1fbb09 : SUCCESS in 9m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f63ba7d774af493880894967e04d6391 : SUCCESS in 7m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa442bd8c1774d0eb68b88a23fa38262 : SUCCESS in 7m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df35b40a31f34564b63c5efe59da9d2c : SUCCESS in 5m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5a7d4dc1d83f4826966cc1e939bafe29 : SUCCESS in 7m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4051a6f486674314a8d3b4e8e59c154b : SUCCESS in 9m 33s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/37e704114a4e4858b6386dc538785b5e : SUCCESS in 1h 19m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0466ea7bc84c4f78881df5ea19396529 : SUCCESS in 7m 21s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/070495ff55e94baea88c4fb99708d136 : SUCCESS in 19m 14s","accounts_in_message":[],"_revision_number":4},{"id":"aee46e5ca7371e9bd44af8c15d03eaae3734b3c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 20:31:07.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ac7cdb14b566a559e37cccce35ae1a20520fb974","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 21:50:50.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a87da51ade04042a4998d589ec64794 : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4a0b58de8b74718a1725e4d32c97fd2 : SUCCESS in 8m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36b800442ebf4db4a40f92c76769097d : FAILURE in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32c4d754ea7c4463aa7cef1ef4da58d0 : SUCCESS in 8m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f3f4330cb321425aadd49dd22df92db9 : SUCCESS in 7m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15247476f4c748bfbdf0402ea7503e86 : SUCCESS in 10m 03s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/55822e4571334693aa57f033b9a7a500 : SUCCESS in 1h 18m 48s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/13fb72f65c644ef6b09e07e303388bd6 : SUCCESS in 18m 15s","accounts_in_message":[],"_revision_number":4},{"id":"910e65e171609acc214067b4a774bdfce6675510","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-03 05:10:28.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"3e0d638b382912623a42f8d955d2d65011171149","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 06:37:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/155f331f7a96409da537704cd3def342 : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/404c07b752a14a56b54e7461a53530fd : SUCCESS in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6fef4d8b64a64b7b874299c482b05eef : SUCCESS in 8m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e051b5361f314245ab045a80031770e6 : SUCCESS in 5m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4d6f6c506bb94d5383888e2553c2f776 : SUCCESS in 7m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff9aefe9aba64b61aa8bf0d412333585 : SUCCESS in 10m 00s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/2cf8ffe08cf14499913801daee3a400e : SUCCESS in 1h 22m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c15e583126c46eea882a741adb74226 : SUCCESS in 7m 45s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/ef034308ce70423286ffa9b121e5631e : SUCCESS in 17m 29s","accounts_in_message":[],"_revision_number":4},{"id":"402810008143466a6e9b052826b44761806cbd2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 06:38:03.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5012152222ba627e68918b4a918ffac339435325","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 08:14:44.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7003a1169bb5414bb2848eee3acf3c57 : SUCCESS in 8m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a753c378ff6d4404beb60e2403163d68 : SUCCESS in 8m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd29a44be8574bd1aae48977aa6878c6 : FAILURE in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f2f098d5b2c41eea1daf798ae1169f6 : SUCCESS in 5m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/81f1cb2778134315a3a95ad220053800 : SUCCESS in 7m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/904312531b404e25aef8f9c0bd5b040b : SUCCESS in 10m 46s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/e72e0ec091a34139a7f5542255a5659d : SUCCESS in 1h 31m 40s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/8cc26823ad9d43e79bca1b66cbe215ba : SUCCESS in 18m 57s","accounts_in_message":[],"_revision_number":4},{"id":"4bbd7a6a793ad5dc39c69b4fde328d004873e0e5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-03 08:15:38.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"898d4623afb3f44bc287dc5f410e4ae6859e73cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 09:58:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecabce30871e497588ed924f955d9d17 : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/583ada2e76a346668182c45b1a4f6253 : SUCCESS in 8m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6428c5d06ac4544abdf84bf8193c9ae : SUCCESS in 9m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb1e9852b75640daa87e40968439b544 : SUCCESS in 8m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ba26b7d2a90e47968d599ed371b4f39c : SUCCESS in 8m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a703379bdab24ab0908f198ca513a33a : SUCCESS in 10m 44s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/95c3f35f13a84c4d95698819f79f53c2 : SUCCESS in 1h 31m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f52e1e6a1092444a8fe375e0b53fe855 : SUCCESS in 9m 27s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/78f8a4a8d6d94678ab4a48b57f75ce12 : SUCCESS in 19m 21s","accounts_in_message":[],"_revision_number":4},{"id":"ce9fba3cbcc854608b34fc3e243fbda33047233a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 09:59:11.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a6a9f04c278a4d64884aca97be45c8ea39c74ef8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 11:27:10.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16fc40efa4be431bbf9ce9d9ca44d297 : SUCCESS in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aac3ad4473c54df081939d6ad8d037b3 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d9dee88239543cd95f128c92bed8d20 : FAILURE in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b02afc0c5624a13a08034a306af5e10 : SUCCESS in 5m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/72a6cb2eeb564257bb3fecd1867e1531 : SUCCESS in 7m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/266327dec6d14150b5d8ed74b71a7dbd : SUCCESS in 9m 35s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/5c440aba185342c091435748ee0511d2 : SUCCESS in 1h 23m 27s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/b3e063d8f8e240539355cb40bc46426a : SUCCESS in 17m 27s","accounts_in_message":[],"_revision_number":4},{"id":"7f1688bfc2742152bb3eec17556ec3259202e0af","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-03 11:41:02.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"7d4fe8ea42ff26af46cbcf8d130e5030bb118607","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 13:02:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d9318e903554d14b226be0c73dcb340 : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a4088d2a6e744c89320650f3477dc64 : SUCCESS in 7m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3df8babed254bb790e9563d9f4cd06d : SUCCESS in 8m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d58bce8008b746158d4d2a5b52d542a3 : SUCCESS in 6m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cd6a62bfdd2c4a62acea1481722b202d : SUCCESS in 7m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e3df348947b4550a7cd63b41eb06f44 : SUCCESS in 9m 54s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/c02ee061c8384dc58d5d553a596b3a6b : SUCCESS in 1h 20m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b09bffbbb4384f01b3191abb651900ae : SUCCESS in 8m 09s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/751468106b1f4a9eafefd9c82531e987 : SUCCESS in 19m 13s","accounts_in_message":[],"_revision_number":4},{"id":"9c94337cc12259886a586f185f1590a90e0c443e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 13:02:21.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5140fafece66d7ceac3b6796c819b8539a333fd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 14:30:41.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/680655d2d0de4044b60cf542ec0f21c8 : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa0090c09a0b420f8ba2ca473c366c91 : SUCCESS in 8m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23efa0a443654d89a5ad2e9e1a1643a8 : FAILURE in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/609f1d32ee7e4aa7835747c0a5b1557e : SUCCESS in 5m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4e14c292712246fda3bf38799929309c : SUCCESS in 8m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a0189200f9f4143976facb1137778c5 : SUCCESS in 10m 10s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/5c3ad40f83d24d8997d7268121a439b6 : SUCCESS in 1h 23m 52s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/1d49d3bdb5ee4efeb6d65403f55f3ecb : SUCCESS in 18m 51s","accounts_in_message":[],"_revision_number":4},{"id":"63419754bd71674f7ff3b87040d3f0bc311364bf","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-03 14:33:58.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"b9b7dd850acdda79062e0377e3cad36a1343342a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 15:59:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9550c73bac8c41a697f9258f959b0fb4 : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/565f40bed4c64c6998ae4a3d3099da68 : SUCCESS in 7m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03ce4ddebba0485aa9b7e4f5d59e87de : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/655f81bef73a4341927434ff42bcf949 : SUCCESS in 7m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df9b0e7d28464d7490bcabe4bb22d41a : SUCCESS in 7m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c011006f198a4fc2a6f76135ca28553b : SUCCESS in 10m 34s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/d27210c46e6d43b3bc8b6b96969cca3c : SUCCESS in 1h 24m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41162a64c29e46d1bc7c059b0c299dfc : SUCCESS in 8m 09s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/d1e306df8ce54c7eaf60652c6f566c79 : SUCCESS in 19m 49s","accounts_in_message":[],"_revision_number":4},{"id":"64b17f674e80bf7a707d8a815d60071c730d90af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 15:59:30.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"bcc12afef396fd71a3062dc4bb50ae99b1b5cc29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 17:29:38.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba68d964e8e440f3bf37092db5b543b5 : SUCCESS in 7m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc438c7026a24fc3b57212e66d6b318a : SUCCESS in 8m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fc02c0a5018436fb84c681b713a83b8 : FAILURE in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bdd27f95e38473f8e4cc0d04adc174a : SUCCESS in 6m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/17c21f2cb09e468d9505422ee14c872e : SUCCESS in 7m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e212eee882d443e983e3b4bd4fd2330 : SUCCESS in 10m 02s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/c66e16b4fde8443e987d3e9f6f560f0e : SUCCESS in 1h 29m 40s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/5254aa90091c48aeb30a8753e3b9ea4f : SUCCESS in 23m 01s","accounts_in_message":[],"_revision_number":4},{"id":"4079763b6c68eac9f5517e99e22ecd26eef8fcdd","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-03 17:35:27.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"b2b64774be8a2042bb22e230bb915a71e207ce7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 19:01:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1743b1d758d74d28902b1cea5c8d0fe8 : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca1f41b87d474773a6e2b9a29fa2d941 : SUCCESS in 8m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/856f0553ea1f41898c97653a0389d557 : SUCCESS in 8m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78b847db8d5a4b23b42d999fa233b0ed : SUCCESS in 8m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6a8d662a9b1e4b2dbeccdd263927c098 : SUCCESS in 8m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1564b57715e34fcc80f822b72653a871 : SUCCESS in 10m 49s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/6f301f315dcd49b196d79bf0f543ca26 : SUCCESS in 1h 24m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f8ed9b758644ae9b4ec03c5d1168cb2 : SUCCESS in 8m 57s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/8f99ca5af1c64e48ab7019464a255e9b : SUCCESS in 18m 34s","accounts_in_message":[],"_revision_number":4},{"id":"068f3b9ad784f3bac1e1743e06018336ff13a301","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 19:01:18.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f0a3419fde4fb38b0d60af363cf1c3b08aa2f912","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 20:21:55.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fafe507c9a24cf69ec15bd12df0222b : SUCCESS in 8m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62d61e90171c4ec7b3a34db2f9606a16 : SUCCESS in 7m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aadd9cc56a1c498896efe75a6f227151 : FAILURE in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8fed65f09404418908dc40bcf15a6d7 : SUCCESS in 6m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c8c7b81f83d408187aea87f65575f90 : SUCCESS in 7m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41ce9a1814cc46548de7006c894a6dcc : SUCCESS in 9m 58s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/a59a00e7e5224544853d82dd2895790b : SUCCESS in 1h 19m 50s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/29c83c76ffa34722bb2dd63d0aa5b7ff : SUCCESS in 17m 59s","accounts_in_message":[],"_revision_number":4},{"id":"888e67b382aff7365bf4cf49a42d7ac9a6ec1cf8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-04 05:47:38.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"366c334b421ea6cb162fd4b4f1de6e8aa441efe6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 06:19:18.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c191436befb4a408a48da6a1220fa4c : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aea76be50334457784465ae1fe3f4574 : SUCCESS in 8m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c82e937bf1544d7a05bc305c1e30624 : SUCCESS in 9m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a89268d06d3495c8209cf3332694252 : SUCCESS in 7m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a6f3665cc27a4484b3c24aa677551720 : SUCCESS in 11m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96b04b25d5f740cd8c61dca1c8168f01 : SUCCESS in 12m 38s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/da9dc3dc6f884da1b36074823d47ebc1 : FAILURE in 6m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3fe469ff3d77407d89ee7016c91f6e76 : SUCCESS in 8m 41s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/fd57d63d96044c58992ca939c91b3d98 : SUCCESS in 23m 13s","accounts_in_message":[],"_revision_number":4},{"id":"9d427fa3907b15b371d24065ccbf822154336798","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-04 06:26:28.000000000","message":"Patch Set 4:\n\nrecheck - network problem","accounts_in_message":[],"_revision_number":4},{"id":"53b1dfa278f659f58047a03e3ccf674c53c6e649","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 07:55:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4cac9d63a5a4963b608120e6ca7ddc9 : SUCCESS in 8m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad846ad75e58415da4aa81beaf083f2c : SUCCESS in 9m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc4d417c2f0f4555b0f5a035af7f4715 : SUCCESS in 8m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a06cddb6c624a21a9bca37b211bc28f : SUCCESS in 6m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f6487a5e16d7477793f06729634eec17 : SUCCESS in 9m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a9dbd1e414643119cc3031d1b6b5f00 : SUCCESS in 12m 23s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/4f43cc814c874999beca97128fe7b529 : SUCCESS in 1h 24m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c16f758f73104eecb6fe4546ad7ef045 : SUCCESS in 7m 53s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/545891fc464a4c89bb259ef2bab12a55 : SUCCESS in 22m 41s","accounts_in_message":[],"_revision_number":4},{"id":"0f3803dfdc33a22b9cb5f584f2e94f545da472f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 07:56:03.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"be8adcfee20cd6c0aa3afc88064bffa8aabc045f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 08:25:22.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/922f76ee11394a6cb330a00b94be08db : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e45de72d29f9420295ab0c9a8de44c34 : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d9c9b07b2864b478285a89ed50affb7 : FAILURE in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35893cc0af5b4982bfb5b8410a7f9f30 : SUCCESS in 6m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2a89eec73c7f47a6b98467132fdf89f3 : SUCCESS in 7m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d0f5063bb124258bc2e36c7433d21da : SUCCESS in 10m 29s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/56c9d6cb8d5b45d3ba6223fbba745292 : FAILURE in 7m 23s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/353ae1b3fb0645f7804800caf5cbf5ee : SUCCESS in 20m 08s","accounts_in_message":[],"_revision_number":4},{"id":"cc3522171b8dd50183f4564a493ecbe285908f56","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-05-04 12:47:35.000000000","message":"Patch Set 4:\n\npy38 is failing on:\n\nFailed to import test module: networking_bgpvpn.tests.unit.services.bagpipe.test_bagpipe\nTraceback (most recent call last):\n  File \"/usr/lib/python3.8/unittest/loader.py\", line 436, in _find_test_path\n    module \u003d self._get_module_from_name(name)\n  File \"/usr/lib/python3.8/unittest/loader.py\", line 377, in _get_module_from_name\n    __import__(name)\n  File \"/home/zuul/src/opendev.org/openstack/networking-bgpvpn/networking_bgpvpn/tests/unit/services/bagpipe/test_bagpipe.py\", line 35, in \u003cmodule\u003e\n    from networking_bgpvpn.neutron.services.service_drivers.bagpipe import bagpipe\n  File \"/home/zuul/src/opendev.org/openstack/networking-bgpvpn/networking_bgpvpn/neutron/services/service_drivers/bagpipe/bagpipe.py\", line 37, in \u003cmodule\u003e\n    from networking_bgpvpn.neutron.services.service_drivers.bagpipe \\\n  File \"/home/zuul/src/opendev.org/openstack/networking-bgpvpn/networking_bgpvpn/neutron/services/service_drivers/bagpipe/bagpipe_v2.py\", line 31, in \u003cmodule\u003e\n    from networking_bagpipe.objects import bgpvpn as bgpvpn_objects\n  File \"/home/zuul/src/opendev.org/openstack/networking-bgpvpn/.tox/py38/lib/python3.8/site-packages/networking_bagpipe/objects/bgpvpn.py\", line 22, in \u003cmodule\u003e\n    from neutron.objects import common_types\nImportError: cannot import name \u0027common_types\u0027 from \u0027neutron.objects\u0027 (/home/zuul/src/opendev.org/openstack/networking-bgpvpn/.tox/py38/lib/python3.8/site-packages/neutron/objects/__init__.py)\n\nThis is due to the networking-bgpvpn tox getting a very old networking-bagipe version that still tries to \"import common_types from neutron.objects\" while most of the code base has moved to importing the common_types from neutron-lib.\n\nIt\u0027s unclear for me why we end up in this situation because I thought that thanks to the tox-siblings mechanisms, declaring networking-bagpipe in required-project in .zuul.yaml should be sufficient to get networking-bagpipe\u0027s master in all tox-based jobs.\n\nFor some reason this does not happen here, I don\u0027t know why.\n\n(I realize that we also need to bump up the minimum release of networking-bagpipe that networking-bgpvpn can use in test-requirements.txt - currently 9.0.0, but nothing in the CI should rely on that, I think)","accounts_in_message":[],"_revision_number":4},{"id":"6c211c3f5c20cb2a709c9081d7c21282fc90eb4f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-04 12:52:02.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c154b032541a1be8301ceb4bb61ff8ba5f6ade60","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-04 12:52:52.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7131d99751df39ce60060136ee62e37d18d63803","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-04 12:53:38.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)\n\nAnd we can drop py37, since that is not part of the new victoria jobs","accounts_in_message":[],"_revision_number":5},{"id":"794ab162a54d110284445e40be409489d351fe36","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2020-05-04 12:56:37.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f8d65c6c3d5fcea4fc70d8b11b156c7c8d2a64b0","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-04 13:47:38.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"42769562d3641b6ca46bbbf33c37b046a9580fdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 14:24:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76c10801bed447a49288d4da21f88b8d : SUCCESS in 7m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c32f32595ba24840ba948ae5287a2045 : SUCCESS in 8m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/423af20d0dc84150b95d18b52c5f4239 : SUCCESS in 8m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80343b0a07c14d13b8a7113918119bdb : SUCCESS in 6m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/76ad4275f85445f7b2774d502fe3ec78 : SUCCESS in 6m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39b6095e5d4143699ad80bb883f2ef75 : SUCCESS in 10m 29s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/3c4836f15129477fb9e04e32dc152aa6 : SUCCESS in 1h 26m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9db8d21f937b49a588789312f60e747d : SUCCESS in 8m 08s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/f25056d19b544a12a9c9d97437ba08c2 : SUCCESS in 19m 20s","accounts_in_message":[],"_revision_number":5},{"id":"23f878ce8097b0871dba28941f47a6ae509751a9","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-04 15:04:05.000000000","message":"Patch Set 5:\n\nThe py37 job comes from horizon-non-primary-django-jobs, will be removed by https://review.opendev.org/#/c/725176/","accounts_in_message":[],"_revision_number":5},{"id":"3ea055ea8a59b65797a60f0deadf11a7c843553c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-05-05 07:22:03.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"4559c723cd8f882d25abb3b278eb30bb2ceeeecc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 07:22:18.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"771fb31993a5bbf2c2bc61919f365bb41462d16e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 08:44:40.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e052aac6e524dc595514013af615c1d : SUCCESS in 10m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c63e6974c5bb493c968a55b6284d2093 : SUCCESS in 9m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a826e537070b4f28a51b6be04b67c037 : SUCCESS in 9m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86e125c83d174498b481478ce3700310 : SUCCESS in 8m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49e2e794eb9a45e286c390b0c45297c0 : SUCCESS in 14m 46s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/d27fe331e29e49bf889175b9f1d31674 : SUCCESS in 1h 20m 42s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/d88db5ade09643dd94df51d931556c27 : SUCCESS in 22m 46s","accounts_in_message":[],"_revision_number":5},{"id":"ea8bc8f4fe421082d86b7ae12a0e21e37eb5ccc9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 08:44:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"743cce12f0e309109cebe24f3d85db2181013768","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 08:48:50.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ec5d9afb86343d889a7f1844a5992cb : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"2042381959349f9832f2f68be8e5d76c7d0d5c03","revisions":{"74aadde7de8a463214e6e0590fafb79af4b9c78f":{"kind":"REWORK","_number":1,"created":"2020-04-22 16:06:28.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/40/722040/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/40/722040/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/40/722040/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/40/722040/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/40/722040/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/40/722040/1"}}},"commit":{"parents":[{"commit":"6d6627743e2729a4a58055200c10a0be63212602","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/6d6627743e2729a4a58055200c10a0be63212602"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-22 16:06:25.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-22 16:06:25.000000000","tz":0},"subject":"Add Python3 victoria unit tests","message":"Add Python3 victoria unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for victoria.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I653f3d863bcdbeac6874a6e51e1e86ea16a18c78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/74aadde7de8a463214e6e0590fafb79af4b9c78f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/74aadde7de8a463214e6e0590fafb79af4b9c78f"}]},"branch":"refs/heads/master"},"279b1aa48d8453f5e435da47160df763a759ec5c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-24 14:05:57.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/40/722040/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/40/722040/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/40/722040/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/40/722040/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/40/722040/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/40/722040/2"}}},"commit":{"parents":[{"commit":"284f2f986477f91896bc86c36fdab1cc2b28b09b","subject":"Add mock to test-requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/284f2f986477f91896bc86c36fdab1cc2b28b09b"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-22 16:06:25.000000000","tz":0},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-04-24 14:05:34.000000000","tz":120},"subject":"Add Python3 victoria unit tests","message":"Add Python3 victoria unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for victoria.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I653f3d863bcdbeac6874a6e51e1e86ea16a18c78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/279b1aa48d8453f5e435da47160df763a759ec5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/279b1aa48d8453f5e435da47160df763a759ec5c"}]},"branch":"refs/heads/master"},"0b535e29168833329edb76a33c8e2e3db21ffb1d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-04-24 14:37:46.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/40/722040/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/40/722040/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/40/722040/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/40/722040/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/40/722040/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/40/722040/3"}}},"commit":{"parents":[{"commit":"157022198a7d5f066d3213cf938fe4ef3ac122d0","subject":"Add mock to test-requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/157022198a7d5f066d3213cf938fe4ef3ac122d0"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-22 16:06:25.000000000","tz":0},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-04-24 14:37:46.000000000","tz":0},"subject":"Add Python3 victoria unit tests","message":"Add Python3 victoria unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for victoria.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I653f3d863bcdbeac6874a6e51e1e86ea16a18c78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/0b535e29168833329edb76a33c8e2e3db21ffb1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/0b535e29168833329edb76a33c8e2e3db21ffb1d"}]},"branch":"refs/heads/master"},"ebfc15aea99709c75f9dbc1f3477c3b3fa440e15":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-04-30 08:17:32.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/40/722040/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/40/722040/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/40/722040/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/40/722040/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/40/722040/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/40/722040/4"}}},"commit":{"parents":[{"commit":"1fbfb94bc08859ad2a5398b32b4fe2e64c98724b","subject":"Fix dsvm-functional and cleanup tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/1fbfb94bc08859ad2a5398b32b4fe2e64c98724b"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-22 16:06:25.000000000","tz":0},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-04-30 08:17:32.000000000","tz":0},"subject":"Add Python3 victoria unit tests","message":"Add Python3 victoria unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for victoria.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I653f3d863bcdbeac6874a6e51e1e86ea16a18c78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/ebfc15aea99709c75f9dbc1f3477c3b3fa440e15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/ebfc15aea99709c75f9dbc1f3477c3b3fa440e15"}]},"branch":"refs/heads/master"},"2042381959349f9832f2f68be8e5d76c7d0d5c03":{"kind":"REWORK","_number":5,"created":"2020-05-04 12:52:52.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/40/722040/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/40/722040/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/40/722040/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/40/722040/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/40/722040/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/40/722040/5"}}},"commit":{"parents":[{"commit":"1fbfb94bc08859ad2a5398b32b4fe2e64c98724b","subject":"Fix dsvm-functional and cleanup tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/1fbfb94bc08859ad2a5398b32b4fe2e64c98724b"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-22 16:06:25.000000000","tz":0},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-04 12:52:20.000000000","tz":120},"subject":"Add Python3 victoria unit tests","message":"Add Python3 victoria unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for victoria.\n\nSee also the PTI in governance [1].\n\nManual change:\n- add py38 with required-projects to gate so that this works in both\n  check and gate\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I653f3d863bcdbeac6874a6e51e1e86ea16a18c78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/2042381959349f9832f2f68be8e5d76c7d0d5c03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/2042381959349f9832f2f68be8e5d76c7d0d5c03"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
