)]}'
{"id":"openstack%2Fnetworking-bgpvpn~769657","triplet_id":"openstack%2Fnetworking-bgpvpn~master~If82415e3de75905b2fb732bf893d59d7ab52c107","project":"openstack/networking-bgpvpn","branch":"master","hashtags":[],"change_id":"If82415e3de75905b2fb732bf893d59d7ab52c107","subject":"Update lower-constraints and requirements","status":"ABANDONED","created":"2021-01-07 03:35:40.000000000","updated":"2021-03-12 07:10:25.000000000","total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"dcec30a06bb6ad8d5bfc417ec611bdfe0ad3855d","_number":769657,"virtual_id_number":769657,"owner":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-04 07:20:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":-1,"date":"2021-03-11 12:21:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-07 05:00:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-07 16:09:45.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2021-01-13 16:30:15.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-01-18 13:18:08.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-01-18 13:18:08.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-01-18 13:18:08.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"8f679dca0986a0afc21567120e659174c8a05d2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-01-07 03:35:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b00306645824a102ef76a5a77270219fe68bb0d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 05:00:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d26a95e9f054adbbed557755d013a71 : SUCCESS in 8m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9b20ac8e0fd4b30812bda47a5349d48 : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a37377519002459d81c0138e8f7fef78 : SUCCESS in 7m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0aba6ed68a149c58b9f0fe822b1d094 : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f83b5d64e547459c9633b5107bb05eb8 : SUCCESS in 7m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b057650ed9d741e7a0f85299ffbd4388 : SUCCESS in 2m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44041a94dca6466eb974dc27c6fd504b : FAILURE in 3m 19s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/ca1bf2f6e66a4809a0a0742800ad3b47 : SUCCESS in 1h 20m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc60035d04a14b96a3c7c77293ce4340 : SUCCESS in 7m 23s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/4d945e96933e4c45a5bcb31842199516 : SUCCESS in 28m 05s","accounts_in_message":[],"_revision_number":1},{"id":"1d245fdf4d1ccae3868d39dee87915e61b12a6f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-01-07 05:42:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c90d6c997be44a9926d6a0e1a2b99cfbabe362d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 07:06:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e92dda4a47549b689b04107fe5b4907 : SUCCESS in 8m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/605d6aa8b15e48cfa3f4a7495d5415a3 : SUCCESS in 8m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3b32afe23e14f43a7b22f19da8301d4 : SUCCESS in 10m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88e2bba569044fdb80ed9470899f183f : SUCCESS in 8m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3c07a9f7bab4d548dc08a7360fccf0b : SUCCESS in 5m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/58e846a1c8e24464b18558938e18202a : SUCCESS in 2m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe970ae7d1d64e78916944e5157136b0 : FAILURE in 4m 04s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/ce0cc0dc5298456e9f6afdc487e7b74f : SUCCESS in 1h 22m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d0e573b6a3f46ca8f02e7a0ceb39f9b : SUCCESS in 9m 17s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/70fbb88263604bf482a65ef0fd099e0b : SUCCESS in 23m 30s","accounts_in_message":[],"_revision_number":2},{"id":"5d43f9ecaa12fc8252a75bf688cda1855324b20a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-01-07 07:23:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"489dbb4db001308f197f83b62d7afdbe9d3f1a91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 08:44:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59b4f3dfa2ee43a4af8d217b1612a66c : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7e6989e618b44ffac452acc9ef5bca9 : SUCCESS in 8m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d655444f5d5e45bf83ea8ca1405a060d : SUCCESS in 7m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ed28d4946f94091aa5c60e8a5d95365 : SUCCESS in 8m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f45777757b5f42cfbeb04a1d1a2a7bc3 : SUCCESS in 8m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e341d8091b5f4cea80f2b3bfda92316a : SUCCESS in 3m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a2f7310002e4987a3d79fecd2d3fd59 : FAILURE in 3m 42s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/e6a2c3b4373b47b8b8c16d2a501b731a : SUCCESS in 1h 18m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4069ece5053c410bb8ff2c6c55ab8c1d : SUCCESS in 8m 19s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/221c557d4fa04d88b440344754031a1b : SUCCESS in 26m 57s","accounts_in_message":[],"_revision_number":3},{"id":"59dffb1f55de5a74545c79ef6867dcc1688df1da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-01-07 09:00:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"cd2c3cef215a66fbc84fdc2eb8e0dc9c7a89c119","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 10:21:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fef4e8d8c55546ee93a65f515cdf7c9a : SUCCESS in 8m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/857081cb774a48579a1e659e0d1b6d20 : SUCCESS in 8m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78447ba7e3c94b649f89445b289f4414 : SUCCESS in 7m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31732b12d7d14d099b868151d142c8a7 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50e93a59de38469e826403125caf2182 : SUCCESS in 7m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c4a08b2a1c1c4afbaaf1a7ee430905d5 : FAILURE in 3m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd34bf9e776a4e27adc42121416e6b6b : FAILURE in 4m 43s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/9c78cf30a5a843e5bba58bb58e5cf6fd : SUCCESS in 1h 18m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d9e33a57cdd42c3b5e2b4dfe8249d59 : SUCCESS in 8m 52s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/5f748f0198ef4ece8bfc97989e215e3b : SUCCESS in 22m 06s","accounts_in_message":[],"_revision_number":4},{"id":"31a6ac0bd5eb06c678c80b01fee469f306216826","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-01-07 11:50:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"830b97b364721cf80b716c0fa651e9a752fbd64c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 13:10:00.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8867360d6b464c7089713c1c826fbfcb : SUCCESS in 9m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60d94d93ab8349b3a2c68c98dd088edc : SUCCESS in 7m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa0c6d06d24c4a3aacb06997da1c82d3 : SUCCESS in 13m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2729eeca73804e3ca2680c3781e7391a : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/299e0a8320fd4a5bbf66b9ea416f0e52 : SUCCESS in 9m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2aec31d01e3d49a48c3253b81fe2de36 : FAILURE in 4m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/560640e31a9d4ed4bac93f4dccafcf03 : FAILURE in 3m 55s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/764cdcb39d4941a4ba48be5fe7f45604 : SUCCESS in 1h 12m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb4da91ba9a54784aa36a91c44ca2891 : SUCCESS in 11m 43s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/26a6cbd2b57e408f8c96780ba10f6609 : SUCCESS in 21m 50s","accounts_in_message":[],"_revision_number":5},{"id":"62c806c50db11aa17048c457b63a11f46c2a6914","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-07 16:09:45.000000000","message":"Patch Set 5:\n\nThanks for working on this","accounts_in_message":[],"_revision_number":5},{"id":"e24347730de2cecd14ee681205a7db6501fd68e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-01-08 01:09:01.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0970f499d19b00f4d8fc5adf831e65c354e68bd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-01-08 01:34:56.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"704905a89e20221c91489b5224fd1f7ec2f9501d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 02:45:05.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba719c9adc084c5790e2923b3632a5f7 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/522056121c824aa3a2e0beb1ab550251 : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e86010554a02494e8c96e24c82c796cd : SUCCESS in 7m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a9291a920314b5fb233a3055e7e3851 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/147bb5eeeaa24a2b82189ee8ef0e0145 : SUCCESS in 7m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/93c3c34c6ae940b183c4843de54ad7c5 : FAILURE in 2m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3cb36c3fb8d14001865567c00d178015 : FAILURE in 3m 54s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/824bc2fafb054d7d9bf0efe5769f7341 : SUCCESS in 1h 08m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f26e26a88cdf42f0914ca54724de233b : SUCCESS in 8m 27s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/0fbbc6706dbe44d68160ef7f420bf1b2 : SUCCESS in 17m 52s","accounts_in_message":[],"_revision_number":7},{"id":"cd7965a37086d1ec2c3f51887deb3ca675608721","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-01-08 05:46:33.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"e176d1bab5756ce97e4813a589291d063f4bde77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 07:06:51.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe91bbb932c442ca95aa8f46254c4589 : SUCCESS in 7m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8259809e0cfa450ba69c7adb254a7c05 : SUCCESS in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f51e9481d3874b26b058ae0d166d48c0 : SUCCESS in 7m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d7cda6eb5644c4d9d030410e669e595 : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18eb26fb0bac4f40812f0ef3740fb02c : SUCCESS in 6m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d94b44259026433c909e9d5e3cc7439b : FAILURE in 2m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d061d19ea8d146cdb76de7690c424de1 : FAILURE in 4m 03s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/4487208ceac14924b1fb89914a76600d : SUCCESS in 1h 15m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/837e2a177cde4ec2ae1dfe6b762e3776 : SUCCESS in 7m 37s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/cac5c463fd864eeab841faffc83522a5 : SUCCESS in 24m 01s","accounts_in_message":[],"_revision_number":8},{"id":"998077205d08d1a654144de232dbbda87836a1f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-01-08 09:38:09.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"0d7fd6fa7fcae42aefc5876dd62aad90806b2ddb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 10:59:38.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf8d312191a546ccb67a2ee8884c38d1 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c04cd7b89854b599a7c43c52bcec5a2 : SUCCESS in 7m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02026f1a6f5a49549c41d77edc0fe2cc : SUCCESS in 7m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/badade6c32fd44f69d53cb98a68821d9 : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a04db36941974559abd4b1c7b6b3b28d : SUCCESS in 6m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c2f7ce1f572749f1be1481ec14a1bab8 : FAILURE in 3m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/051f3aa887ef4daba20294ebc97c6eba : FAILURE in 4m 02s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/a9692bb4ebec40eea87d97ecddd36676 : SUCCESS in 1h 14m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71fc2752315e4b8082d50bbcc96eee2c : SUCCESS in 8m 57s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/2a3245f2deeb4b1f8ca1a3c644f9bc6e : SUCCESS in 21m 10s","accounts_in_message":[],"_revision_number":9},{"id":"efd9a60b0c538ac2ba3b89f8e9b1655c5c77b7a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-01-11 03:12:51.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"ac289380482f6a5ab384edea225b77ab1f63d12b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 04:19:30.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fefdd7bd12f42f6a8fe51611e80810a : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6b6563ce18043f3923a11497a26e1d1 : SUCCESS in 7m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d74dfa17c6434c988077cfb8c24c0cf4 : SUCCESS in 7m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2987327867494c0ca14f4b163a384051 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/565f4fa18ebc41408f1b230c363ebf0c : SUCCESS in 6m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ead1bd2031fb4274912f160c7e42bdd9 : SUCCESS in 2m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e57af05427e24379b09b9e7afb908d52 : FAILURE in 3m 56s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/40468b0001a042ceaf87ddd546151561 : SUCCESS in 1h 05m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/757513a5be334f279c61003d6b6b4cdb : SUCCESS in 7m 57s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/e8f184b91a1841ee93d1de73a18f2698 : SUCCESS in 16m 48s","accounts_in_message":[],"_revision_number":10},{"id":"a54a9c5016f58ef1262e79b836311484796ea010","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-01-11 05:45:56.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"f73bdf0831a5fb1b51d3aca4fa4883d36ae9adcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 06:55:58.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/210e02d6ed27411abd2468a2e30361fa : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cdbebb0917434c2bb35e019a00c97e78 : SUCCESS in 10m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7a46a81418b485d85f168b80066f379 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6467be828d5543229e83c2e93b5bd482 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cd6c69d1bcf48ca9ebe7c674f09f43e : SUCCESS in 6m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6d1ed251d83245448f1a53404f1f2ddb : SUCCESS in 2m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/828b408b78a04ea7bf2ae1afd7ba7447 : FAILURE in 3m 20s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/548d899cab2b4d75ac616397fa728623 : SUCCESS in 1h 08m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ec78d6ad2694744a6fdf35fbd376cb9 : SUCCESS in 8m 32s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/39fa5f60398549d18c8b877d9c35de77 : SUCCESS in 19m 46s","accounts_in_message":[],"_revision_number":11},{"id":"4b7ab2189cd99f5d810c798bd32f7163a742b4d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-01-11 07:29:48.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"cc2013293439adcb4065e409c24dea49cc741c29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 08:34:12.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b401a14a9514f45ad1ad49301223d1c : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f67706db16f4d3c90d2cde7d16953ca : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64e74c8b8bf84635af0c045a867bdf55 : SUCCESS in 8m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/749f6b70f68a4982a1792b4b0a854225 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1f46e7bafeb4e9082b67eb984fd1ceb : SUCCESS in 7m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c913da42b5014c64a6a860946340596d : SUCCESS in 4m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca5b605dd8f1441b8ce3676d4f2ca056 : FAILURE in 6m 16s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/358a2ea11d3c4e3ba01ce1bc5cd36790 : SUCCESS in 1h 02m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0219e9c0ce24124a9f2bfc3ca77e500 : SUCCESS in 6m 15s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/3e5b1244b45940e59a2b40b656bab8e3 : FAILURE in 37m 00s","accounts_in_message":[],"_revision_number":12},{"id":"cf841ca9aed6a3e60fca29925170c61e5697496f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-01-12 02:21:53.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"04d49c08ae5354c7b8eb3fc7944d9ea49a558d51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 03:52:01.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5116948ca20a4472a702beca2c702fea : SUCCESS in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7009795ab69240e3b237f2caecea40e0 : SUCCESS in 7m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2520e77713b49658e012a266ac64cd7 : SUCCESS in 7m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5bad5504824444ea75fa07a9eb7c192 : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/757158d69a5d4a3bb86a4e35208a7cbc : SUCCESS in 6m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/66dec8b755c2473382c10cc13327f122 : SUCCESS in 2m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b11e2b1237004516a2e71c1def65a067 : FAILURE in 3m 59s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/b6dba161321f4e538a5e39e81fba7a93 : SUCCESS in 1h 27m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d9474c2319e4164a23d5e23dbb4d301 : SUCCESS in 8m 38s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/8cfd1149bdb64a51bfdc55dd671e0be3 : FAILURE in 35m 12s","accounts_in_message":[],"_revision_number":13},{"id":"7dc249d9d390e52b8dbb66b0a774ede0b1c4632f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-01-12 07:43:09.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"54f2207388eececf1ab99a3451991af2f9c76e4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 08:20:57.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d41683b7dc246e49d45696b6bb67c7b : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab89f7f02f5542768bd8f6d246bb8305 : SUCCESS in 10m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa24d2e27f0b4cb6a724acb1760b09d0 : FAILURE in 21m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd55fcd047664feaa3d59465c34c8f1e : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e24c54ddfb24475953582fc5d06424b : SUCCESS in 7m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/982a536eca084b0792d848fe3283ca53 : SUCCESS in 2m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/426efd7ff920459091e9c0e17b8f5edb : FAILURE in 5m 18s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/b5326735101f43ce878076dc7761ebca : FAILURE in 8m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f608c00d002495999b9b9db8910856e : FAILURE in 8m 51s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/a79619c0abd94dec835d28fb39eaaba2 : SUCCESS in 36m 37s","accounts_in_message":[],"_revision_number":14},{"id":"0a3f2f7560d204126c4f4627f75cb846a328dc13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-01-12 08:43:10.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"d8f9e3493a8008dc14baf07789855075777e0be0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 10:11:28.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/033f1c30926f40ce96c57dac38e06d66 : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa591d5ad10240d5bf7459f141ec6c32 : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af2b504d016a4334b89540131531536c : SUCCESS in 8m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/76ca95378eb245589132c3952c560fb2 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0392dc9610fb4b388a9a747bac92ed2f : SUCCESS in 7m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/99599caa38204cf4a29412213bafb8d1 : SUCCESS in 3m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/372d05b2010f4b1a826e1a7d68671b8f : TIMED_OUT in 41m 01s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/c89ec56a9a33463d958b9e30143f9008 : SUCCESS in 1h 25m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f13465d6f26d4c7bb5e350365ac975de : SUCCESS in 9m 48s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/903096a59dbe40498c8bbe77fa3c0aad : SUCCESS in 30m 02s","accounts_in_message":[],"_revision_number":15},{"id":"17100974df305cdee433383c39ab1f490b10f823","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-01-12 10:14:05.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"792109e632aec3d22b7a565727c45e2ef10be008","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 11:35:04.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e4c6ac6265a47cf9336ee6036e38273 : SUCCESS in 9m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76290b42d1dd4a0cb5eb41a0dbadc12d : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7148f248f1dc414597ca877eb20741ff : SUCCESS in 18m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44a1e7f67adf4cfeb8c7e0165bd6559b : SUCCESS in 9m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0d46cc2246c4927abd279a0795aaea1 : SUCCESS in 12m 35s\n- requirements-check https://zuul.opendev.org/t/openstack/build/817f8212f2b548ff915e08be9af96a34 : SUCCESS in 4m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7774e47e997b49408445f49575846765 : TIMED_OUT in 41m 02s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/77b8a2d68d614928872dcb3f5a81ca1d : SUCCESS in 1h 16m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/035f4ec91dd2462b978d05e1d0419203 : SUCCESS in 7m 39s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/48f7cbdf7dea4ca58ac2e695029dfc1b : SUCCESS in 26m 27s","accounts_in_message":[],"_revision_number":15},{"id":"ca7f2518d885880d2403108e46a66c126c29371e","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-01-13 01:57:34.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"9bd63bba624c286085182a784530b2f7c2f6ba77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 03:21:56.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92aec67c5a234dbe9c03d000203914e3 : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7b73898f32d47f6bad7b33123fe475e : SUCCESS in 8m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f2637488e394c42a4e694c3b09ea0c2 : SUCCESS in 9m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc50ddabb044459ca554f1a1a9bf3b17 : SUCCESS in 8m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc004f110bb0426dae0369a9fa12f5e7 : SUCCESS in 7m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/092a7fc3c7b44814b271cfcf0c40f43a : SUCCESS in 2m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3cbc69bcf50040dfb214a91b277877c1 : TIMED_OUT in 40m 48s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/264322b0d1794d0db03f74ea8cc196be : SUCCESS in 1h 21m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f67c29b0e5cd40019ae2e85ce95ef626 : SUCCESS in 10m 01s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/b46abfafa72541ed8be3f569bba8e6cb : SUCCESS in 21m 14s","accounts_in_message":[],"_revision_number":15},{"id":"12efbe103aba524e36542bc87a3827de4f1595d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-13 16:30:15.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"56e39f4dde3dd48d3732790408307868b8c47a32","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-13 16:30:53.000000000","message":"Patch Set 16:\n\nlet\u0027s see if this helps","accounts_in_message":[],"_revision_number":16},{"id":"2cf514e253cda59331c6befe8080aec13737fba0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 17:52:22.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cdf6631305e40bbb989b9d05b8218dc : SUCCESS in 7m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28a31e5f15ee4c908874984d9993dae1 : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8280c5bcc9b4ded893c1a479b2d677a : SUCCESS in 10m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1bb89cf547b46c3bd239e7d7823f90f : SUCCESS in 8m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33fe7037586647e59c4ef524c01d9d51 : SUCCESS in 6m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/10774cfd41b345ee95b6112d7ff84777 : FAILURE in 3m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e21570f5b3864d1183d9ad8c0567a639 : FAILURE in 5m 48s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/9235450cc7014a80a84e2c6c63dc869c : SUCCESS in 1h 13m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3ce50254bfd46f3b12b0c098e92a50d : SUCCESS in 10m 12s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/a24f31d3d044426cab5da53e3d5f80aa : SUCCESS in 22m 50s","accounts_in_message":[],"_revision_number":16},{"id":"f0d1d586a2cd78ff34d2f5d035678800bb94b605","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-01-14 02:14:02.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"dc36ae8748fad88dc512d41ec4717d942fe4dab1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 03:51:07.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70279f707d6d4d1a9277be64292af7f7 : SUCCESS in 10m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d854b303f2b4445eba3fd54833db1c5e : SUCCESS in 10m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d392e859b2948f783aa99982869b00c : SUCCESS in 10m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f54cf944920648758e63d5a0b62f3934 : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72f8f0e222494a0d833919217c69772b : SUCCESS in 10m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2d655264bef24b999bccf311605dfa1b : SUCCESS in 3m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e82dda4cd48449f889a7cf960ec4f436 : TIMED_OUT in 41m 28s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/729a13222ab9476fade8996eba703b98 : SUCCESS in 1h 30m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88668d522a064e69947844f3b831f625 : SUCCESS in 8m 23s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/3abb4c75ba6a4408bcc6366aa0b2ab02 : SUCCESS in 22m 02s","accounts_in_message":[],"_revision_number":17},{"id":"0bd4cb762ce82d69adc6da7e7480f9b5c3585e61","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-01-14 06:26:48.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"658f640323107984d06b00710dccd6b966a2c700","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 08:17:28.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ab0ed55cb6446ac896098ba59827365 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/daffccebf35546bb91f9a039e1e2b9ab : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ea046809a84486bbf89feae9369575b : SUCCESS in 7m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d80a9431f3f84b98bb2d6bd32f034209 : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdd7cab33dac4889b785bb6340ce1210 : SUCCESS in 6m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/58eb651731c146a5a4bda921a2928d9e : SUCCESS in 3m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa18aff2703e48b3a205f0272bd696fc : TIMED_OUT in 41m 26s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/60c405edd16f42be929fbd5a0267b8f8 : SUCCESS in 1h 24m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3d34a2102bc4ac88a6617c7e6d05dfa : SUCCESS in 7m 36s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/8426dcc947784bdbbf490c559566a942 : SUCCESS in 18m 57s","accounts_in_message":[],"_revision_number":17},{"id":"219f4a0210b8bcc7da4015c290b56e1316c3b5f8","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-01-14 09:46:37.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"db7de7d0c55af48aa71061f096c70e89230418de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 11:39:35.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f6db4acd3b848afbdc54c4aaffa20ba : SUCCESS in 7m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3068e7642ab452f8c73489fd7302dcc : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d610a47fd34e47288b20279c3971cdb8 : SUCCESS in 6m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a0e25c7d4e940a28a9a78e38fd52742 : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/905d794497ed4cde9b76fc1d6e1945a4 : SUCCESS in 10m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/267f75f019294514a69fff7c092e7aec : SUCCESS in 2m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d1984daba764ba6bc7fa2188ce72463 : TIMED_OUT in 40m 50s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/2a86945aaeb044e8b08a1886184a1743 : SUCCESS in 1h 16m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/613aac3163d84fefa38d8e79d5e28d94 : SUCCESS in 8m 17s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/74ccc6fc9beb461fa2a3c44d63950283 : SUCCESS in 22m 28s","accounts_in_message":[],"_revision_number":17},{"id":"e80807b9476bb04031ef61e90c8641678aa2737c","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-01-15 03:12:57.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"df4e6c26bf6804209eb80af253b4f409672e6bcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 04:33:40.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4d3c20d21c840649081f2fc51e4fc8b : SUCCESS in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d766d9395b643aea07c6d7efa60ec13 : SUCCESS in 8m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d38062b2c2c48afad9e8e6ddfe17007 : SUCCESS in 10m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e6704262c8e471a86272d57a5c82259 : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3534e39bb81f41f9a2fd9e320b77522d : SUCCESS in 6m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/932c3019296543cc956f22dda6bb8ec1 : SUCCESS in 3m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/839cfbdaaafe484ab609811ae8f8f062 : TIMED_OUT in 40m 59s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/30b461d7fd874b9ca5b80d54d45831d6 : SUCCESS in 1h 16m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fae09387ac441e7888de549a44db7ac : SUCCESS in 8m 30s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/9a9ff104676d4b19bd701a8adf810bdd : SUCCESS in 23m 55s","accounts_in_message":[],"_revision_number":17},{"id":"9f5d81ea9240e11da6aaad2c2487902fd40353da","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-01-18 13:16:40.000000000","message":"Patch Set 17:\n\nTo make life easier I pushed a patch to make lower-constraints job non-voting:\nhttps://review.opendev.org/c/openstack/networking-bgpvpn/+/771219","accounts_in_message":[],"_revision_number":17},{"id":"b4042076fc6d4962bb1fcb96735a5cde59a34450","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-03-03 09:47:07.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"263ce1f42e18aad8f23a34dd82ec497bee34eee9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 11:16:52.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/548085e32d2644a1b551088904d1080c : SUCCESS in 10m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/059643199ccf49dd8819304cc624d467 : SUCCESS in 10m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3a36538f0a34b37afd95487f27c4eea : SUCCESS in 11m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/549ab7bb6ea4482b9430277780db534b : SUCCESS in 9m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0d7e863481141c9920c728601df7bc0 : SUCCESS in 10m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/29d4d97df38d427799e028f521913192 : SUCCESS in 4m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/979979b39f14439dac81c280d61c42f4 : TIMED_OUT in 41m 36s (non-voting)\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/9b1580dc1f364adcb2aea23d975b4f97 : SUCCESS in 1h 18m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9ed7a150aa64a73bc1e74ac128c5eda : SUCCESS in 10m 27s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/3a46f11d79744a67a7f32c04db359540 : SUCCESS in 24m 28s","accounts_in_message":[],"_revision_number":17},{"id":"35b3cc5b8ce8126da4182588be6d1c1252c33770","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-03 12:18:54.000000000","message":"Patch Set 17: Code-Review-1\n\n(2 comments)\n\nThanks for working on this, please check my comments","accounts_in_message":[],"_revision_number":17},{"id":"c15810f3e857882eb94755ce51def8552422cc21","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-03-04 05:58:22.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"23c39957ee0d1acb78d462e2fb672dc260f2d738","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-03-04 05:58:26.000000000","message":"Patch Set 18: Commit message was updated.","accounts_in_message":[],"_revision_number":18},{"id":"d400185342cec14c5377038367ef33e749526085","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 07:20:02.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3c209f963a24a7f871787abb6b81194 : SUCCESS in 9m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69245fb60ba04e1aa2fa7a2754ac970c : SUCCESS in 9m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a69f4a179211482eac5552376699d372 : SUCCESS in 10m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/288d80feb5e649fba256a08a7c22d128 : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d06fe277b274366a62596ac64749e8e : SUCCESS in 7m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/203524f007a84d80a2cab69b87a0d2de : SUCCESS in 4m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a36f267790164a2683a8fe0d2f19027a : TIMED_OUT in 44m 21s (non-voting)\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/fcc05d4ff03c428286e139ef97f4d770 : SUCCESS in 1h 20m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d62ca7dd46884f4fbf40e3e0906255d9 : SUCCESS in 9m 38s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/0e1d6d507c9b49668a1c8eaf3fdab93f : SUCCESS in 22m 57s","accounts_in_message":[],"_revision_number":18},{"id":"b057a9107a965b35bf0fa0b94781da7a321d5b02","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-04 11:16:46.000000000","message":"Patch Set 18:\n\nIf your goal is to fix lower-constraints job it\u0027s timeouted (it\u0027s now non-voting(:\nhttps://b35cb237d2390b546d7d-2bd0a90c56ef0ba69bd0efa56ba93067.ssl.cf2.rackcdn.com/769657/18/check/openstack-tox-lower-constraints/a36f267/job-output.txt\n\nNot an expert on pip resolver issues but this timeout seems related to some hard to follow dependency chain.","accounts_in_message":[],"_revision_number":18},{"id":"f46506d3fb4c4ce6e5158762938ac9b5b9983cf6","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-03-08 03:12:45.000000000","message":"Patch Set 18:\n\n\u003e Patch Set 18:\n\u003e \n\u003e If your goal is to fix lower-constraints job it\u0027s timeouted (it\u0027s now non-voting(:\n\u003e https://b35cb237d2390b546d7d-2bd0a90c56ef0ba69bd0efa56ba93067.ssl.cf2.rackcdn.com/769657/18/check/openstack-tox-lower-constraints/a36f267/job-output.txt\n\u003e \n\u003e Not an expert on pip resolver issues but this timeout seems related to some hard to follow dependency chain.\n\nI know that submitting multiple tests in a patch is not good, but I don’t have a local environment for tox testing. I will pay attention next time. \nActually I want to fix the lower-constraints job in this one https://review.opendev.org/c/openstack/networking-bgpvpn/+/769634.","accounts_in_message":[],"_revision_number":18},{"id":"c78d6de19b379000ba66211e78912c97115a3dd0","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-03-10 11:15:29.000000000","message":"Patch Set 18:\n\nYou can make small venv to see how lower constraints go with new pip resolver with a command like:\npython3.8 -m venv foo \u0026\u0026  foo/bin/pip install -U pip \u0026\u0026 foo/bin/pip install -c lower-constraints.txt -r test-requirements.txt -r requirements.txt \n\nFailures are then easy to fix/spot, timeouts can be helped a lot by updating minimum versions on packages that generate a lot of backtrack. Checking the other neutron and stadium projects can be useful too as most changes will be similar","accounts_in_message":[],"_revision_number":18},{"id":"ccf0202452c9d16654f8ffc24f93ab624a436a19","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-11 12:21:03.000000000","message":"Patch Set 18: Code-Review-1\n\n\u003e Patch Set 18:\n\u003e \n\u003e \u003e Patch Set 18:\n\u003e \u003e \n\u003e \u003e If your goal is to fix lower-constraints job it\u0027s timeouted (it\u0027s now non-voting(:\n\u003e \u003e https://b35cb237d2390b546d7d-2bd0a90c56ef0ba69bd0efa56ba93067.ssl.cf2.rackcdn.com/769657/18/check/openstack-tox-lower-constraints/a36f267/job-output.txt\n\u003e \u003e \n\u003e \u003e Not an expert on pip resolver issues but this timeout seems related to some hard to follow dependency chain.\n\u003e \n\u003e I know that submitting multiple tests in a patch is not good, but I don’t have a local environment for tox testing. I will pay attention next time. \n\u003e Actually I want to fix the lower-constraints job in this one https://review.opendev.org/c/openstack/networking-bgpvpn/+/769634.\n\nThe patch https://review.opendev.org/c/openstack/networking-bgpvpn/+/769634 is green now, so this is not necessary anymore","accounts_in_message":[],"_revision_number":18},{"id":"dcec30a06bb6ad8d5bfc417ec611bdfe0ad3855d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-03-12 07:10:25.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"cda43c85e3a69d447157545c7d0dae27d20346c4","revisions":{"36ff2e2c2225abd33fcce2874b04095594e445e0":{"kind":"REWORK","_number":1,"created":"2021-01-07 03:35:40.000000000","uploader":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"ref":"refs/changes/57/769657/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/57/769657/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/1"}}},"commit":{"parents":[{"commit":"2722d5a0ad8e921064b875b00c5108afd2937676","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/2722d5a0ad8e921064b875b00c5108afd2937676"}]}],"author":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-07 03:34:57.000000000","tz":480},"committer":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-07 03:34:57.000000000","tz":480},"subject":"Update lower-constraints","message":"Update lower-constraints\n\nChange-Id: If82415e3de75905b2fb732bf893d59d7ab52c107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/36ff2e2c2225abd33fcce2874b04095594e445e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/36ff2e2c2225abd33fcce2874b04095594e445e0"}]},"branch":"refs/heads/master"},"e848ffece50368bcf8a8df300b079e8f24405860":{"kind":"REWORK","_number":2,"created":"2021-01-07 05:42:04.000000000","uploader":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"ref":"refs/changes/57/769657/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/57/769657/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/2"}}},"commit":{"parents":[{"commit":"2722d5a0ad8e921064b875b00c5108afd2937676","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/2722d5a0ad8e921064b875b00c5108afd2937676"}]}],"author":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-07 03:34:57.000000000","tz":480},"committer":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-07 05:41:35.000000000","tz":480},"subject":"Update lower-constraints","message":"Update lower-constraints\n\nChange-Id: If82415e3de75905b2fb732bf893d59d7ab52c107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/e848ffece50368bcf8a8df300b079e8f24405860"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/e848ffece50368bcf8a8df300b079e8f24405860"}]},"branch":"refs/heads/master"},"97d461612443ec21fa9bf9a56d79be83afea89ac":{"kind":"REWORK","_number":3,"created":"2021-01-07 07:23:13.000000000","uploader":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"ref":"refs/changes/57/769657/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/57/769657/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/3"}}},"commit":{"parents":[{"commit":"2722d5a0ad8e921064b875b00c5108afd2937676","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/2722d5a0ad8e921064b875b00c5108afd2937676"}]}],"author":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-07 03:34:57.000000000","tz":480},"committer":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-07 07:22:37.000000000","tz":480},"subject":"Update lower-constraints","message":"Update lower-constraints\n\nChange-Id: If82415e3de75905b2fb732bf893d59d7ab52c107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/97d461612443ec21fa9bf9a56d79be83afea89ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/97d461612443ec21fa9bf9a56d79be83afea89ac"}]},"branch":"refs/heads/master"},"8f745736e7390fb1a8937f84e9d778cf390882ec":{"kind":"REWORK","_number":4,"created":"2021-01-07 09:00:59.000000000","uploader":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"ref":"refs/changes/57/769657/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/57/769657/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/4"}}},"commit":{"parents":[{"commit":"2722d5a0ad8e921064b875b00c5108afd2937676","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/2722d5a0ad8e921064b875b00c5108afd2937676"}]}],"author":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-07 03:34:57.000000000","tz":480},"committer":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-07 09:00:20.000000000","tz":480},"subject":"Update lower-constraints","message":"Update lower-constraints\n\nChange-Id: If82415e3de75905b2fb732bf893d59d7ab52c107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/8f745736e7390fb1a8937f84e9d778cf390882ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/8f745736e7390fb1a8937f84e9d778cf390882ec"}]},"branch":"refs/heads/master"},"e30117b42f51b3f5cad31f431e7a33a0d8fad24a":{"kind":"REWORK","_number":5,"created":"2021-01-07 11:50:31.000000000","uploader":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"ref":"refs/changes/57/769657/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/57/769657/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/5"}}},"commit":{"parents":[{"commit":"2722d5a0ad8e921064b875b00c5108afd2937676","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/2722d5a0ad8e921064b875b00c5108afd2937676"}]}],"author":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-07 03:34:57.000000000","tz":480},"committer":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-07 11:49:57.000000000","tz":480},"subject":"Update lower-constraints","message":"Update lower-constraints\n\nChange-Id: If82415e3de75905b2fb732bf893d59d7ab52c107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/e30117b42f51b3f5cad31f431e7a33a0d8fad24a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/e30117b42f51b3f5cad31f431e7a33a0d8fad24a"}]},"branch":"refs/heads/master"},"4c1bd88cbe83e4d5ac573bfefdca34ab2e96acc4":{"kind":"REWORK","_number":6,"created":"2021-01-08 01:09:01.000000000","uploader":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"ref":"refs/changes/57/769657/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/57/769657/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/6"}}},"commit":{"parents":[{"commit":"2722d5a0ad8e921064b875b00c5108afd2937676","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/2722d5a0ad8e921064b875b00c5108afd2937676"}]}],"author":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-07 03:34:57.000000000","tz":480},"committer":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-08 01:08:29.000000000","tz":480},"subject":"Update lower-constraints","message":"Update lower-constraints\n\nChange-Id: If82415e3de75905b2fb732bf893d59d7ab52c107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/4c1bd88cbe83e4d5ac573bfefdca34ab2e96acc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/4c1bd88cbe83e4d5ac573bfefdca34ab2e96acc4"}]},"branch":"refs/heads/master"},"5642eb119df5a2dc77d29df53465bc84b6a12523":{"kind":"REWORK","_number":7,"created":"2021-01-08 01:34:56.000000000","uploader":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"ref":"refs/changes/57/769657/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/57/769657/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/7"}}},"commit":{"parents":[{"commit":"2722d5a0ad8e921064b875b00c5108afd2937676","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/2722d5a0ad8e921064b875b00c5108afd2937676"}]}],"author":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-07 03:34:57.000000000","tz":480},"committer":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-08 01:34:26.000000000","tz":480},"subject":"Update lower-constraints","message":"Update lower-constraints\n\nChange-Id: If82415e3de75905b2fb732bf893d59d7ab52c107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/5642eb119df5a2dc77d29df53465bc84b6a12523"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/5642eb119df5a2dc77d29df53465bc84b6a12523"}]},"branch":"refs/heads/master"},"5acd4454bb7bb03a3cfe6ad2d7da80ecb0ca9c90":{"kind":"REWORK","_number":8,"created":"2021-01-08 05:46:33.000000000","uploader":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"ref":"refs/changes/57/769657/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/57/769657/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/8"}}},"commit":{"parents":[{"commit":"2722d5a0ad8e921064b875b00c5108afd2937676","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/2722d5a0ad8e921064b875b00c5108afd2937676"}]}],"author":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-07 03:34:57.000000000","tz":480},"committer":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-08 05:45:54.000000000","tz":480},"subject":"Update lower-constraints","message":"Update lower-constraints\n\nChange-Id: If82415e3de75905b2fb732bf893d59d7ab52c107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/5acd4454bb7bb03a3cfe6ad2d7da80ecb0ca9c90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/5acd4454bb7bb03a3cfe6ad2d7da80ecb0ca9c90"}]},"branch":"refs/heads/master"},"b3b06cafe369cdb10959c70c3beac1506ccd8319":{"kind":"REWORK","_number":9,"created":"2021-01-08 09:38:09.000000000","uploader":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"ref":"refs/changes/57/769657/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/57/769657/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/9"}}},"commit":{"parents":[{"commit":"2722d5a0ad8e921064b875b00c5108afd2937676","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/2722d5a0ad8e921064b875b00c5108afd2937676"}]}],"author":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-07 03:34:57.000000000","tz":480},"committer":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-08 09:37:41.000000000","tz":480},"subject":"Update lower-constraints","message":"Update lower-constraints\n\nChange-Id: If82415e3de75905b2fb732bf893d59d7ab52c107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/b3b06cafe369cdb10959c70c3beac1506ccd8319"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/b3b06cafe369cdb10959c70c3beac1506ccd8319"}]},"branch":"refs/heads/master"},"6b3de61101349beb3bfb0f2143ec048b8e7632ae":{"kind":"REWORK","_number":10,"created":"2021-01-11 03:12:51.000000000","uploader":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"ref":"refs/changes/57/769657/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/57/769657/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/10"}}},"commit":{"parents":[{"commit":"2722d5a0ad8e921064b875b00c5108afd2937676","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/2722d5a0ad8e921064b875b00c5108afd2937676"}]}],"author":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-07 03:34:57.000000000","tz":480},"committer":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-11 03:12:17.000000000","tz":480},"subject":"Update lower-constraints","message":"Update lower-constraints\n\nChange-Id: If82415e3de75905b2fb732bf893d59d7ab52c107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/6b3de61101349beb3bfb0f2143ec048b8e7632ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/6b3de61101349beb3bfb0f2143ec048b8e7632ae"}]},"branch":"refs/heads/master"},"2c7ecec8e0d5fee2d5c6c5355a3fb28aa15da518":{"kind":"REWORK","_number":11,"created":"2021-01-11 05:45:56.000000000","uploader":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"ref":"refs/changes/57/769657/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/57/769657/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/11"}}},"commit":{"parents":[{"commit":"2722d5a0ad8e921064b875b00c5108afd2937676","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/2722d5a0ad8e921064b875b00c5108afd2937676"}]}],"author":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-07 03:34:57.000000000","tz":480},"committer":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-11 05:45:26.000000000","tz":480},"subject":"Update lower-constraints","message":"Update lower-constraints\n\nChange-Id: If82415e3de75905b2fb732bf893d59d7ab52c107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/2c7ecec8e0d5fee2d5c6c5355a3fb28aa15da518"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/2c7ecec8e0d5fee2d5c6c5355a3fb28aa15da518"}]},"branch":"refs/heads/master"},"73888dfb4effbbdaabc225f5be6b509a9ce8f5a7":{"kind":"REWORK","_number":12,"created":"2021-01-11 07:29:48.000000000","uploader":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"ref":"refs/changes/57/769657/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/57/769657/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/12"}}},"commit":{"parents":[{"commit":"2722d5a0ad8e921064b875b00c5108afd2937676","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/2722d5a0ad8e921064b875b00c5108afd2937676"}]}],"author":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-07 03:34:57.000000000","tz":480},"committer":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-11 07:29:16.000000000","tz":480},"subject":"Update lower-constraints","message":"Update lower-constraints\n\nChange-Id: If82415e3de75905b2fb732bf893d59d7ab52c107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/73888dfb4effbbdaabc225f5be6b509a9ce8f5a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/73888dfb4effbbdaabc225f5be6b509a9ce8f5a7"}]},"branch":"refs/heads/master"},"515f870db299f27e3768c505359730fb994fbfe9":{"kind":"REWORK","_number":13,"created":"2021-01-12 02:21:53.000000000","uploader":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"ref":"refs/changes/57/769657/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/57/769657/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/13"}}},"commit":{"parents":[{"commit":"2722d5a0ad8e921064b875b00c5108afd2937676","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/2722d5a0ad8e921064b875b00c5108afd2937676"}]}],"author":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-07 03:34:57.000000000","tz":480},"committer":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-12 02:21:23.000000000","tz":480},"subject":"Update lower-constraints","message":"Update lower-constraints\n\nChange-Id: If82415e3de75905b2fb732bf893d59d7ab52c107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/515f870db299f27e3768c505359730fb994fbfe9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/515f870db299f27e3768c505359730fb994fbfe9"}]},"branch":"refs/heads/master"},"805ef1660c4675b8e50d37856796096a56c80a54":{"kind":"REWORK","_number":14,"created":"2021-01-12 07:43:09.000000000","uploader":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"ref":"refs/changes/57/769657/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/57/769657/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/14"}}},"commit":{"parents":[{"commit":"2722d5a0ad8e921064b875b00c5108afd2937676","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/2722d5a0ad8e921064b875b00c5108afd2937676"}]}],"author":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-07 03:34:57.000000000","tz":480},"committer":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-12 07:42:29.000000000","tz":480},"subject":"Update lower-constraints","message":"Update lower-constraints\n\nChange-Id: If82415e3de75905b2fb732bf893d59d7ab52c107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/805ef1660c4675b8e50d37856796096a56c80a54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/805ef1660c4675b8e50d37856796096a56c80a54"}]},"branch":"refs/heads/master"},"92301858450527b98eb740203723532acfa2bd0b":{"kind":"REWORK","_number":15,"created":"2021-01-12 08:43:10.000000000","uploader":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"ref":"refs/changes/57/769657/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/57/769657/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/15"}}},"commit":{"parents":[{"commit":"2722d5a0ad8e921064b875b00c5108afd2937676","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/2722d5a0ad8e921064b875b00c5108afd2937676"}]}],"author":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-07 03:34:57.000000000","tz":480},"committer":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-12 08:42:46.000000000","tz":480},"subject":"Update lower-constraints","message":"Update lower-constraints\n\nChange-Id: If82415e3de75905b2fb732bf893d59d7ab52c107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/92301858450527b98eb740203723532acfa2bd0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/92301858450527b98eb740203723532acfa2bd0b"}]},"branch":"refs/heads/master"},"b68f490a064ad320206f9888bd06be23bbce9263":{"kind":"REWORK","_number":16,"created":"2021-01-13 16:30:15.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/57/769657/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/57/769657/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/16"}}},"commit":{"parents":[{"commit":"2722d5a0ad8e921064b875b00c5108afd2937676","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/2722d5a0ad8e921064b875b00c5108afd2937676"}]}],"author":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-07 03:34:57.000000000","tz":480},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-01-13 16:29:52.000000000","tz":60},"subject":"Update lower-constraints","message":"Update lower-constraints\n\nChange-Id: If82415e3de75905b2fb732bf893d59d7ab52c107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/b68f490a064ad320206f9888bd06be23bbce9263"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/b68f490a064ad320206f9888bd06be23bbce9263"}]},"branch":"refs/heads/master"},"ecae7410258ba45094a9f82c99aa2b693b9ec986":{"kind":"REWORK","_number":17,"created":"2021-01-14 02:14:02.000000000","uploader":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"ref":"refs/changes/57/769657/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/57/769657/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/17"}}},"commit":{"parents":[{"commit":"2722d5a0ad8e921064b875b00c5108afd2937676","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/2722d5a0ad8e921064b875b00c5108afd2937676"}]}],"author":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-07 03:34:57.000000000","tz":480},"committer":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-14 02:05:36.000000000","tz":480},"subject":"Update lower-constraints","message":"Update lower-constraints\n\nChange-Id: If82415e3de75905b2fb732bf893d59d7ab52c107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/ecae7410258ba45094a9f82c99aa2b693b9ec986"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/ecae7410258ba45094a9f82c99aa2b693b9ec986"}]},"branch":"refs/heads/master"},"cda43c85e3a69d447157545c7d0dae27d20346c4":{"kind":"NO_CODE_CHANGE","_number":18,"created":"2021-03-04 05:58:26.000000000","uploader":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"ref":"refs/changes/57/769657/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/57/769657/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/57/769657/18"}}},"commit":{"parents":[{"commit":"2722d5a0ad8e921064b875b00c5108afd2937676","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/2722d5a0ad8e921064b875b00c5108afd2937676"}]}],"author":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-07 03:34:57.000000000","tz":480},"committer":{"name":"wdd","email":"wangzihao@yovole.com","date":"2021-03-04 05:58:08.000000000","tz":0},"subject":"Update lower-constraints and requirements","message":"Update lower-constraints and requirements\n\nUpdate lower-constraints and requirements to newer versions for pass ci jobs.\nChange-Id: If82415e3de75905b2fb732bf893d59d7ab52c107\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/cda43c85e3a69d447157545c7d0dae27d20346c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/cda43c85e3a69d447157545c7d0dae27d20346c4"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
