)]}'
{"id":"openstack%2Fnetworking-bgpvpn~790064","triplet_id":"openstack%2Fnetworking-bgpvpn~stable%2Ftrain~I2b5fcb89fef2345fe35068407c885af038b420ab","project":"openstack/networking-bgpvpn","branch":"stable/train","topic":"disable_lc","hashtags":[],"change_id":"I2b5fcb89fef2345fe35068407c885af038b420ab","subject":"WIP: make l-c job non-voting","status":"ABANDONED","created":"2021-05-06 13:29:25.000000000","updated":"2021-05-07 16:02:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1e01407f949771fddc2986919e0391d0e1fdc84f","_number":790064,"virtual_id_number":790064,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-05-06 14:39:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-06 14:39:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aa401bd735450147e6a06e2a4b451745ffd42657","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-05-06 13:29:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d047d700f9996f1ccb76c065839974c068517216","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 14:39:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aedf7a37c73549579a6123d5bfd8bbdc : FAILURE in 7m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c56e452bcdc9412fa10b10c269e8e783 : SUCCESS in 8m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9cb5dbe90bab4450857332e5f40e6b66 : SUCCESS in 8m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/747c4aeedb7c4310b5ac90b4e699c265 : SUCCESS in 9m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60c0338fcb08499b813d8fbd1288cfa6 : SUCCESS in 8m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/069de63aaa0441f686c579c3d478091f : FAILURE in 6m 46s (non-voting)\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/39e278ac6c75436794f97dc05c31f5bd : SUCCESS in 8m 40s\n- neutron-tempest-plugin-bgpvpn-bagpipe-train https://zuul.opendev.org/t/openstack/build/607124825cd941f2aa530c32f55ebfba : SUCCESS in 1h 08m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/015e5077ffde41bebbeab209514d7280 : SUCCESS in 9m 16s\n- networking-bgpvpn-dsvm-functional https://zuul.opendev.org/t/openstack/build/65065d4ce0d443b08b4b72dc652b13ed : FAILURE in 18m 46s","accounts_in_message":[],"_revision_number":1},{"id":"1e01407f949771fddc2986919e0391d0e1fdc84f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-05-07 16:02:24.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"18cca729794516c3109dd5cfec2395bd0a8223a0","revisions":{"18cca729794516c3109dd5cfec2395bd0a8223a0":{"kind":"REWORK","_number":1,"created":"2021-05-06 13:29:25.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/64/790064/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/64/790064/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/64/790064/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/64/790064/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/64/790064/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/64/790064/1"}}},"commit":{"parents":[{"commit":"909ade03794c8dc169ff81488ba4047bc5e06262","subject":"Revert \"[stable-only] Cap horizon to Train version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/909ade03794c8dc169ff81488ba4047bc5e06262"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-05-06 13:28:45.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-05-06 13:28:45.000000000","tz":120},"subject":"WIP: make l-c job non-voting","message":"WIP: make l-c job non-voting\n\nChange-Id: I2b5fcb89fef2345fe35068407c885af038b420ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/18cca729794516c3109dd5cfec2395bd0a8223a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/18cca729794516c3109dd5cfec2395bd0a8223a0"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
