)]}'
{"id":"openstack%2Fnetworking-bgpvpn~796447","triplet_id":"openstack%2Fnetworking-bgpvpn~stable%2Fstein~I775af9673e8fdbc70b439d92b355bbc00cd27432","project":"openstack/networking-bgpvpn","branch":"stable/stein","topic":"clean-non-voting-jobs-stable/stein","hashtags":[],"change_id":"I775af9673e8fdbc70b439d92b355bbc00cd27432","subject":"[EM releases] Move non-voting jobs to the experimental queue","status":"ABANDONED","created":"2021-06-15 14:49:23.000000000","updated":"2021-08-15 08:30:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":796476,"cherry_pick_of_patch_set":1,"meta_rev_id":"f2ca095a7b5468106fac4f42895f6216d553932c","_number":796447,"virtual_id_number":796447,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-06-15 16:11:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2021-06-16 07:42:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-18 06:43:49.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-15 16:11:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-16 07:42:55.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-06-18 06:43:49.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"}],"messages":[{"id":"7401f696a7ecef2c6c11d3a5989b4785783c41d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-15 14:49:23.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"146095abaf4e2e489be66a766a26f1d31d05dd2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 16:11:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5137fcc9cbb458fb79ba8096e7d1b52 : FAILURE in 9m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/612f2c1e88c54332a81e2fbb3ee164a8 : SUCCESS in 7m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f90bf148ff4a436fbeaffd3f64bd7c50 : SUCCESS in 10m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d7128a870b446499a7dd83d1c60836b : SUCCESS in 9m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9574be78eae945f1b8e52f4d41445240 : SUCCESS in 10m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/802180af37494f2389cfe19ace4a1ea4 : SUCCESS in 5m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ead739ba7e940d0a12364dc1580ff8e : FAILURE in 5m 04s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/487288a7f29343c8bd782e00edecde5e : SUCCESS in 6m 18s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/9cca81c4f3b44f2bba1e133a94fc6b58 : SUCCESS in 1h 17m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72ab56e040a64f099511606f5b662eae : SUCCESS in 9m 05s\n- networking-bgpvpn-dsvm-functional https://zuul.opendev.org/t/openstack/build/700f8de319dd4f76964a0ea3bf3c070b : FAILURE in 17m 54s","accounts_in_message":[],"_revision_number":1},{"id":"568dc8398e6aca5a7f4c02a6dd792a92a78823e1","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-06-16 07:42:55.000000000","message":"Patch Set 1: Code-Review+2\n\nChange itself LGTM though that branch needs some fixes apparently for pep8, l-c and functional","accounts_in_message":[],"_revision_number":1},{"id":"3458c8e4cfbce7b04dcd6a9729ec4c045ce02994","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2021-06-18 06:43:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f2ca095a7b5468106fac4f42895f6216d553932c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-15 08:30:14.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"983c4b19453a0fba114937dd7ddbec9444324f12","revisions":{"983c4b19453a0fba114937dd7ddbec9444324f12":{"kind":"REWORK","_number":1,"created":"2021-06-15 14:49:23.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/47/796447/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/47/796447/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/47/796447/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/47/796447/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/47/796447/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/47/796447/1"}}},"commit":{"parents":[{"commit":"0e0ee5a8e7ef0271764f2e7cfa9b2411fdaf34da","subject":"Use tox \u0027all\u0027 env instead of \u0027all-plugins\u0027 for networking bgpvpn tempest plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/0e0ee5a8e7ef0271764f2e7cfa9b2411fdaf34da"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-15 13:48:46.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-15 14:49:23.000000000","tz":0},"subject":"[EM releases] Move non-voting jobs to the experimental queue","message":"[EM releases] Move non-voting jobs to the experimental queue\n\nDuring the CI meeting we agreed that non-voting jobs in the branches\nwhich are in Extened Maintenance (EM) phase should be moved to from\nthe check to the experimental queue.\n\nThis patch is doing exactly that.\n\nChange-Id: I775af9673e8fdbc70b439d92b355bbc00cd27432\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/983c4b19453a0fba114937dd7ddbec9444324f12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/983c4b19453a0fba114937dd7ddbec9444324f12"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
