)]}'
{"id":"openstack%2Fnetworking-bgpvpn~806101","triplet_id":"openstack%2Fnetworking-bgpvpn~master~I240355d5a0ebf8ece14bb2b4faf764c23477e322","project":"openstack/networking-bgpvpn","branch":"master","topic":"assertItemsEqual","hashtags":[],"change_id":"I240355d5a0ebf8ece14bb2b4faf764c23477e322","subject":"Use assertCountEqual instead of assertItemsEqual","status":"MERGED","created":"2021-08-26 08:17:02.000000000","updated":"2021-08-27 12:39:01.000000000","submitted":"2021-08-27 12:37:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806101-assertItemsEqual","meta_rev_id":"3d245b94d1de72724f81ecb572aa740c30d19f42","_number":806101,"virtual_id_number":806101,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-27 12:37:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-08-27 11:05:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-08-27 11:19:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2021-08-27 08:35:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-27 10:02:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-08-27 11:19:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-26 09:24:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-26 14:48:29.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-08-26 14:48:29.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-08-27 05:17:05.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-08-27 05:17:05.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2021-08-27 05:17:05.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-08-27 10:02:42.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"}],"messages":[{"id":"be233eedabff7bdffc04f73e72dfbbb4e5c9db49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-08-26 08:17:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56dcf2a9c3094e7f17f9b34288e77eea7047dfa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 09:24:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18b1a6194fc0431297161c352f941e60 : FAILURE in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1acc72adad1f4a9a860ab4e3764e1027 : SUCCESS in 10m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55d0a3c1dd7a440b9724656f3596482c : FAILURE in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f1903d6db9744bd954c5be86494f3b9 : SUCCESS in 9m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a35d827e958a4752bc8ef2d8d1f5c2d3 : SUCCESS in 7m 42s\n- horizon-tox-python3-django30 https://zuul.opendev.org/t/openstack/build/747962f12d9441d69bce4cf8bf30069b : FAILURE in 9m 11s (non-voting)\n- horizon-tox-python3-django31 https://zuul.opendev.org/t/openstack/build/a2701a48e82543ff9776eb57203117cf : FAILURE in 10m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c97dbe79ab7e48f4a3041bb77597ec5e : FAILURE in 4m 56s (non-voting)\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/de86b76ba04c45dcaa2ad0f43fb93148 : SUCCESS in 1h 05m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7886f6c8bcdc4b4186c639af69d501b2 : FAILURE in 5m 43s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/8351fab9448345118a0ff7c298affa55 : SUCCESS in 22m 04s","accounts_in_message":[],"_revision_number":1},{"id":"9b26d328e7404662a517e75015bc02c49bc64935","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-08-26 14:48:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bc6a448bddced8c25ca012d930129cf3d8f6f3d6","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-08-26 14:48:53.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e recheck\n\npypi issue resolved","accounts_in_message":[],"_revision_number":1},{"id":"e8f2eb5ba973ba4175a0cf85d9a0b53bd2601604","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 15:53:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54db2f92cb0849f6a6b5ace881d9c631 : SUCCESS in 11m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61c669d3d93346b3915fe28fd47494ad : SUCCESS in 11m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2877c32ebdc54e8b830243512d0a4c7d : SUCCESS in 11m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/baad9e8e0eef4ed49e361134386c756e : SUCCESS in 9m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94e63a3b0bbd417fa73197ee56d612cf : SUCCESS in 8m 43s\n- horizon-tox-python3-django30 https://zuul.opendev.org/t/openstack/build/5ce840c654f64eecbe88613244d33974 : FAILURE in 12m 25s (non-voting)\n- horizon-tox-python3-django31 https://zuul.opendev.org/t/openstack/build/dfde7c70d4fa442ebb7ae2fdca3f4ce6 : FAILURE in 10m 46s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8c3cae224e54f6baff91be027421b80 : FAILURE in 6m 47s (non-voting)\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/7fd396c41a9a4b0b8fa30e6b59e620a8 : SUCCESS in 1h 02m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e726f8d2efab4c649fee19a5d628bcf9 : SUCCESS in 10m 13s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/e71a2c70a2c742f381d1a38adcd3a841 : SUCCESS in 26m 36s","accounts_in_message":[],"_revision_number":1},{"id":"bc58b433da86a1c55425ebd4431d10d0753bbcf4","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-08-27 08:35:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4c41488883897d268b1db6eb18bfc2ef45885c15","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2021-08-27 10:02:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"94c2187a6b83973f9bc3449d052b4fd777506647","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-08-27 11:05:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ef9df80f5787970c410699f40ce528054602db79","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-27 11:19:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7030d41718652b941fa10475212d3663732cb13c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 11:20:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4853d667b2ee1d4757f100f1dfcd4a3492b719c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 12:37:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8805296d2faf4d95a30124d64836c735 : SUCCESS in 9m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92a4f471d6bd44c5bb853ea2208bdd8e : SUCCESS in 10m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbcecf4b162743b0a2764ade5c95a2a0 : SUCCESS in 8m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc78e3ed35384be4932c033de7efe27d : SUCCESS in 8m 50s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/db0a6f853d3d4ae4be7b1e642f6c5592 : SUCCESS in 1h 15m 02s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/3538d36211764dc4835ae4bb98348424 : SUCCESS in 22m 48s","accounts_in_message":[],"_revision_number":1},{"id":"8dfb86fd9581d77debc11bea914b41aa274acc98","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 12:37:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3d245b94d1de72724f81ecb572aa740c30d19f42","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 12:39:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7d5356db7ee47a1a15a81c1a7daf8bf : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78fc8e35cc92ede933f21e1ed35608047b5b6aa9","revisions":{"78fc8e35cc92ede933f21e1ed35608047b5b6aa9":{"kind":"REWORK","_number":1,"created":"2021-08-26 08:17:02.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/01/806101/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/01/806101/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/01/806101/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/01/806101/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/01/806101/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/01/806101/1"}}},"commit":{"parents":[{"commit":"4267d257d858c04e83fae97f63ca1525ea83ef5d","subject":"Merge \"Changed minversion in tox to 3.18.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/4267d257d858c04e83fae97f63ca1525ea83ef5d"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-08-26 08:16:08.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-08-26 08:16:58.000000000","tz":330},"subject":"Use assertCountEqual instead of assertItemsEqual","message":"Use assertCountEqual instead of assertItemsEqual\n\nThe assertItemsEqual method has been removed in Python 3.3 [1] but\nit was kept alive by unittest2, imported by testtools. For better\ncompatibility, change to assertCountEqual.\n\n[1] https://bugs.python.org/issue17866\n\nChange-Id: I240355d5a0ebf8ece14bb2b4faf764c23477e322\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/78fc8e35cc92ede933f21e1ed35608047b5b6aa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/78fc8e35cc92ede933f21e1ed35608047b5b6aa9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
