)]}'
{"id":"openstack%2Fnetworking-bgpvpn~809499","triplet_id":"openstack%2Fnetworking-bgpvpn~stable%2Frocky~I9efad8c23ddee86150b4da4c178209224b9c65db","project":"openstack/networking-bgpvpn","branch":"stable/rocky","topic":"remove-c7-jobs-stable/rocky","attention_set":{},"removed_from_attention_set":{"20182":{"account":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"last_update":"2021-12-15 18:14:11.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I9efad8c23ddee86150b4da4c178209224b9c65db","subject":"Remove jobs corresponds to obselete featuresets","status":"ABANDONED","created":"2021-09-17 06:12:20.000000000","updated":"2021-12-15 18:14:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":809524,"cherry_pick_of_patch_set":1,"meta_rev_id":"284e4794e72fc4530e8a45d42d142d3eb8bce9fa","_number":809499,"virtual_id_number":809499,"owner":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-17 08:12:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-17 06:12:41.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-17 08:12:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-15 18:14:11.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"23cf0302f327f1a9659b595e574d3b51e767bf7f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-17 06:12:20.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.\n\nThe following files contain Git conflicts:\n* .zuul.yaml","accounts_in_message":[],"_revision_number":1},{"id":"4da6f19cdc897dc58d5f602f138b65445a0a2a51","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-09-17 06:12:41.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/networking-bgpvpn on branch stable/rocky.  The error was:\n\n  while scanning a simple key\n  in \"openstack/networking-bgpvpn/.zuul.yaml@stable/rocky\", line 57, column 1\ncould not find expected \u0027:\u0027\n  in \"openstack/networking-bgpvpn/.zuul.yaml@stable/rocky\", line 58, column 9","accounts_in_message":[],"_revision_number":1},{"id":"d2f4d39359354986835381cbe91c35f253509f80","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-17 06:13:04.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"713c366a1ecb656c8271390786df5c826912138e","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-17 06:13:15.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"45d952013fb351a5f996c06a3da17ffd6c6994c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 08:12:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7ae9aa10b5348bdbbe89c7ae7b461af : SUCCESS in 10m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/14e050465e1b463baece9437bed85bf2 : SUCCESS in 9m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b417aaebea2e46ccb8ef1ae5eac605ff : FAILURE in 6m 06s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ccff434555724df396d56eeb569b010e : SUCCESS in 7m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b75cf23937343208a217c5c247b90c1 : FAILURE in 5m 46s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/3517136854af4a47924daf5f17e9b598 : FAILURE in 6m 14s\n- legacy-tempest-dsvm-networking-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/547b6412a62f420cbc70175ceaab91b4 : SUCCESS in 1h 26m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fb06ae42fe64d71a9e7f9b88e778e36 : SUCCESS in 10m 07s\n- networking-bgpvpn-dsvm-functional https://zuul.opendev.org/t/openstack/build/029a64b8b5e34741b742ead12b059437 : FAILURE in 18m 02s","accounts_in_message":[],"_revision_number":2},{"id":"284e4794e72fc4530e8a45d42d142d3eb8bce9fa","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-12-15 18:14:11.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ca7d455dbf34bf9e12a07f3d537c0697c282f6da","revisions":{"5cc82f1d8044990e0fbfbda23d28d7dc5016e05d":{"kind":"REWORK","_number":1,"created":"2021-09-17 06:12:20.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/99/809499/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/99/809499/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/99/809499/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/99/809499/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/99/809499/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/99/809499/1"}}},"commit":{"parents":[{"commit":"18d1cf6928ca276b0dab2f849aea293e3a8f42e7","subject":"Fix lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/18d1cf6928ca276b0dab2f849aea293e3a8f42e7"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 06:11:07.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 06:12:20.000000000","tz":0},"subject":"Remove jobs corresponds to obselete featuresets","message":"Remove jobs corresponds to obselete featuresets\n\nAs we are removing the jobs which are using the obselete featuresets [1]\nso removing those jobs from templates/layouts.\nAnd this branch is also EOL for TripleO.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: I9efad8c23ddee86150b4da4c178209224b9c65db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/5cc82f1d8044990e0fbfbda23d28d7dc5016e05d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/5cc82f1d8044990e0fbfbda23d28d7dc5016e05d"}]},"branch":"refs/heads/stable/rocky"},"ca7d455dbf34bf9e12a07f3d537c0697c282f6da":{"kind":"REWORK","_number":2,"created":"2021-09-17 06:13:04.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/99/809499/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/99/809499/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/99/809499/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/99/809499/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/99/809499/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/99/809499/2"}}},"commit":{"parents":[{"commit":"18d1cf6928ca276b0dab2f849aea293e3a8f42e7","subject":"Fix lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/18d1cf6928ca276b0dab2f849aea293e3a8f42e7"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 06:11:07.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 06:12:59.000000000","tz":0},"subject":"Remove jobs corresponds to obselete featuresets","message":"Remove jobs corresponds to obselete featuresets\n\nAs we are removing the jobs which are using the obselete featuresets [1]\nso removing those jobs from templates/layouts.\nAnd this branch is also EOL for TripleO.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: I9efad8c23ddee86150b4da4c178209224b9c65db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/ca7d455dbf34bf9e12a07f3d537c0697c282f6da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/ca7d455dbf34bf9e12a07f3d537c0697c282f6da"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
