)]}'
{"id":"openstack%2Fnetworking-bgpvpn~809501","triplet_id":"openstack%2Fnetworking-bgpvpn~stable%2Fwallaby~Id5bc15a22921a34399e926736d0a9a478598aff7","project":"openstack/networking-bgpvpn","branch":"stable/wallaby","topic":"c7-to-c8-stable/wallaby","attention_set":{},"removed_from_attention_set":{"20182":{"account":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"last_update":"2021-10-11 10:22:52.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2021-10-11 07:39:14.000000000","reason":"Slawek Kaplonski replied on the change"}},"hashtags":[],"change_id":"Id5bc15a22921a34399e926736d0a9a478598aff7","subject":"Run TripleO jobs on CentOS8 instead of CentOS7","status":"MERGED","created":"2021-09-17 06:16:25.000000000","updated":"2021-10-11 10:36:21.000000000","submitted":"2021-10-11 10:22:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809501-c7-to-c8-stable/wallaby","cherry_pick_of_change":809525,"cherry_pick_of_patch_set":1,"meta_rev_id":"08187e3858e2da013b98e540bd00a96ea715a05b","_number":809501,"virtual_id_number":809501,"owner":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-11 10:22:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-10-11 07:39:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-10-11 07:37:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2021-10-11 07:39:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-17 08:54:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-06 07:21:13.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-10-11 07:37:42.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"57b3ba013c4f0cd7e29d4250f1eb709a9469a5fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-17 06:16:25.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5b437a813eadc3e78babc207f0fecf8e82bf4a3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 08:54:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb4ea1b89c6d43a1a607e57540aaf377 : SUCCESS in 9m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd4067a4026944d8aaa9f91ea0624b25 : SUCCESS in 11m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/382423110f4a4361818ebb1df5553ab3 : SUCCESS in 8m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/312237a020dd4155bf842ab8419c57e5 : SUCCESS in 10m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89b868b0ba784689bc71095ba9374337 : SUCCESS in 10m 29s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/e34d7340261e45bc8abb137f9c072f28 : SUCCESS in 1h 39m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/51b8ffa8fbba49c899dec95e1e5c5ffe : FAILURE in 4m 34s (non-voting)\n- neutron-tempest-plugin-bgpvpn-bagpipe-wallaby https://zuul.opendev.org/t/openstack/build/30a76fff4c2f447aa2aa67adcde42a68 : SUCCESS in 1h 11m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94a342f55fde408e97a5bf4073e2833a : SUCCESS in 10m 53s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/079b89bf66724ae9b64f223379e0233d : SUCCESS in 24m 03s","accounts_in_message":[],"_revision_number":1},{"id":"14c4164c3b617aabad4664fcf990f802bd4814b4","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-10-06 07:21:13.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"59c9379c2cbbfce15079365d0b809a670591865a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-10-11 07:09:33.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c84a4d5b25e37ac6c13d1319b771ac85470e926a","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-10-11 07:10:50.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"98425fa236d3aaad7a2684ec17ba2b9c0d6b73c7","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-10-11 07:37:42.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1028215a361e0dee82d80d4a51a1695f30af9c67","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-10-11 07:39:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ecdddf3944e97b73901a1a4fde5c76d543b946a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 09:00:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5f0ae75cfb740489c4de39826655224 : SUCCESS in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4c70ac40d6d4d1ea92f8d1ff94ca903 : SUCCESS in 10m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3ce9e0d54f342c292032d6b0af441cf : SUCCESS in 8m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/998be5b01b694e1ab7a750783e3cbcee : SUCCESS in 9m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f050b9be2cbe40b2b593285a7760adad : SUCCESS in 8m 05s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/09a78be84afe4b7bb92dff515d77e2b3 : SUCCESS in 1h 42m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20c6ea7d9431447aacee969b0845bacd : FAILURE in 5m 36s (non-voting)\n- neutron-tempest-plugin-bgpvpn-bagpipe-wallaby https://zuul.opendev.org/t/openstack/build/010075a0a8474736ba5e29717d468218 : SUCCESS in 1h 15m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/345d7931451f4a76b2e2e76d72081624 : SUCCESS in 10m 57s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/b0a49ddc21e74b05a11a5dc183ea3768 : SUCCESS in 30m 10s","accounts_in_message":[],"_revision_number":2},{"id":"d519cf26a2a67494bd1c6a7db34c09ccf1ee394d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 09:01:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"746613037e4f879e58173ec62a5d149fcb9f2641","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 10:22:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55e9209b32f0452699858ce1a59c3ed2 : SUCCESS in 7m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48cb87060ed44b80b98488f313d08ad9 : SUCCESS in 10m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2fadc042f0ea427193ba93cfac4aefe8 : SUCCESS in 8m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3a52c4236494dad9047280f6f9f30a9 : SUCCESS in 7m 02s\n- neutron-tempest-plugin-bgpvpn-bagpipe-wallaby https://zuul.opendev.org/t/openstack/build/63624059fde648a6a0d0ef6d0ead2600 : SUCCESS in 1h 17m 47s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/e7c7e1dad9ab49ed88cf0bdacac8c669 : SUCCESS in 32m 58s","accounts_in_message":[],"_revision_number":2},{"id":"ec753721fc4d330a97eea6f50b08f8bc92ff0d64","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 10:22:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"08187e3858e2da013b98e540bd00a96ea715a05b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 10:36:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f21b7148d7b4bdfa5954ddfaf278db9 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"188121ae61f16656b721485147e6feb98201bb99","revisions":{"19977fda0920e6e774f247f67f944f104cceb468":{"kind":"REWORK","_number":1,"created":"2021-09-17 06:16:25.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/01/809501/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/01/809501/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/01/809501/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/01/809501/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/01/809501/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/01/809501/1"}}},"commit":{"parents":[{"commit":"b070a43b69221b4a4967329b0fa34044a07f24b7","subject":"Use assertCountEqual instead of assertItemsEqual","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/b070a43b69221b4a4967329b0fa34044a07f24b7"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 06:14:37.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 06:16:25.000000000","tz":0},"subject":"Run TripleO jobs on CentOS8 instead of CentOS7","message":"Run TripleO jobs on CentOS8 instead of CentOS7\n\nAs we are cleaning up the c7 jobs and obselete featuresets [1].\nThis change replaces usage of CentOS7 in TripleO jobs by CentOS8.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: Id5bc15a22921a34399e926736d0a9a478598aff7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/19977fda0920e6e774f247f67f944f104cceb468"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/19977fda0920e6e774f247f67f944f104cceb468"}]},"branch":"refs/heads/stable/wallaby"},"188121ae61f16656b721485147e6feb98201bb99":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-10-11 07:09:33.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/01/809501/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/01/809501/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/01/809501/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/01/809501/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/01/809501/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/01/809501/2"}}},"commit":{"parents":[{"commit":"b070a43b69221b4a4967329b0fa34044a07f24b7","subject":"Use assertCountEqual instead of assertItemsEqual","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/b070a43b69221b4a4967329b0fa34044a07f24b7"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 06:14:37.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-10-11 07:09:33.000000000","tz":0},"subject":"Run TripleO jobs on CentOS8 instead of CentOS7","message":"Run TripleO jobs on CentOS8 instead of CentOS7\n\nAs we are cleaning up the c7 jobs and obselete featuresets [1].\nThis change replaces usage of CentOS7 in TripleO jobs by CentOS8.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: Id5bc15a22921a34399e926736d0a9a478598aff7\n(cherry picked from commit e403f297b02906641a040969e1d5b62a08f62cf2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/188121ae61f16656b721485147e6feb98201bb99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/188121ae61f16656b721485147e6feb98201bb99"}]},"branch":"refs/heads/stable/wallaby","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
